)]}'
{"id":"openstack%2Fmonasca-common~763707","triplet_id":"openstack%2Fmonasca-common~master~Id2a9678b311af32bc6ec9a9832ce423042868e35","project":"openstack/monasca-common","branch":"master","hashtags":[],"change_id":"Id2a9678b311af32bc6ec9a9832ce423042868e35","subject":"Add py38 in setup metadata","status":"MERGED","created":"2020-11-23 02:12:53.000000000","updated":"2020-12-07 12:07:24.000000000","submitted":"2020-12-07 12:07:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763707","meta_rev_id":"42abaa133e30fbab37669d3799ee489b63890083","_number":763707,"virtual_id_number":763707,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-07 12:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-07 11:23:34.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-07 11:23:34.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 02:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 19:55:11.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"CC"},{"updated":"2020-12-07 11:23:34.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"473296d686fcdedbe29ab6f9726a1ae3c35fd310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-11-23 02:12:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6fcfbf84cd09de2fa47ef1c71010887569202c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 02:56:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1edcc9fbde774278b12a9542f6a8d123 : SUCCESS in 1m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/562c90e17df4451b85d161e832485b21 : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/710e36f5e8424507a4ef5017d973bbed : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/265082792f094a028435f0a87df991b4 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/278dc20a85e84185a769b76f4ac4e1ba : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffbbaf1ca5964287873524a80b337839 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7e017030d9a4e7596ed07fe5ad2a667 : SUCCESS in 3m 24s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e992b632b6f745a9a46f7135b856fc10 : FAILURE in 41m 40s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/f4b98ce6662241efb27a288ac99e9df5 : FAILURE in 33m 37s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/5d97b91b0a06408c86c6c70b3f3e59af : SUCCESS in 1m 47s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/9f06595570384391a0ee66a40e597c99 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a03735a237ff9cc489148a060f86c646f40bfe4a","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-11-23 03:29:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"821b9b97f3bacd0fe9649ace39e428632790c57c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 04:09:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a73618a7e7940a4821d597264332c96 : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf3c4d17fff0468396584f6b206a2a6a : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68e43a820d2f49c18e56a8cd99725107 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/407acdfc28f548f0a56d293def9ec198 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e1aff0b4fc0404bbb5dc15347728eb8 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/449ffc52140c4ca59272156f3eadf218 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef7bdecb7a7d4fefb6ebf6036350c4e5 : SUCCESS in 3m 43s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/361f3397d55340b290d4addde94eb897 : FAILURE in 36m 30s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/f8ceff514cf94f19825fb840fd979bf5 : FAILURE in 34m 02s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/e1bbfdec58ce494a876edc37df556fbc : SUCCESS in 2m 02s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/ca860a3edc734992a8f38383341c5493 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"cfb43e4c632f4e50e8357b78a9db2e5454bb7917","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-11-23 08:29:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ad160d6b8e84877b54eb0c2de154eb2433ae326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 10:29:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f3e4afbe74384903afbcdaae0d7a09fa : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/055eea9ecbeb4cbc97a2ef3de70a336b : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/296cd0daab404cd1b7e57aff2f6ab06a : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb11adfad3d402496875e1e556e13c7 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6f1e781b87f4aabb13d6b5538d8db9e : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bae946b9bc3442b19417684ef1c8ad6b : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aec373db1414ef78dcfb3a42372d59e : SUCCESS in 5m 01s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/48866e5234424c4996c632091396a3de : FAILURE in 40m 59s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/8c305693eb1142668cd8b411bb3e955a : FAILURE in 35m 48s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/cfaaed716cb448788f78380db771d48b : SUCCESS in 2m 32s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/9a4fe30363ce4084bc8c6429a80f6fed : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"155434f0f3740ea8356c6ddc42213c426e7d81dd","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-11-25 02:15:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"48e714ee516732d4f02dc700e35fd4c518af56ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 03:28:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9c20bc530e6f4e989913a8bec36e94ec : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f1b793274f34706b41d45b12328179f : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91ae34d424ca4bfbbe70816c81552a1d : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d450256fdda48bd8a0458234574b5ab : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2580947d2cb64c3398c0094c6bfd86e5 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d852cce050646a29a10ca111e5fd752 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/639d0c86ef8c45bd8df0c9b7f1ff0203 : SUCCESS in 3m 54s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/eac6b2bb6df54ff9b4d21b08ed6130b4 : FAILURE in 34m 40s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/9322bca441764a86b85b141512484320 : FAILURE in 54m 50s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/12d06a2669294d10a124c95f483f868f : SUCCESS in 2m 00s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/6952d4a4599e48d0a3d32b47863a3bd7 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ee814d9e42792561b021b06475fbeb116dad94ac","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-12-04 01:47:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb43d7000257ddab263372996b7051adb1ced356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 03:13:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/234c6c5371eb472b8760128ff140d309 : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f6c01be48ba49069f91698615d2aece : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7a2b3d325b744bbaa6e45abd5879b57 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef06316d32594e98b45ce246fae412ba : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eac81184fe5b49eaa2719cc8dbf18bd9 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14ce2136ccf44ba691616749951be7c7 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e76fd93a76441b9a41c1739ec1b8854 : SUCCESS in 3m 55s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/b74fe446b3084b4e934207ed19cd4206 : FAILURE in 57m 14s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/147156e05fe14ee6a7df4052b36ac3ff : FAILURE in 34m 09s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/baf3bfc3244340349458db6e7b7ccef1 : SUCCESS in 2m 16s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/685480ed3e7747c2b10a0bc2531e3bb5 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c430667da60fc50542d1486f4c0c0460b8da71f6","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-12-04 19:55:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78edce29729a7fedf0723bebe66c95089bb64a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 20:45:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/95f938bdf785412891c36ea8ed06bcfc : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ee6d2df614943d4b652eacb17882e41 : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/264478ba5c8742b682b1f96099dcf88f : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71454a7b8aa24d5d9878aa6c8b81d4b3 : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54890070fb5b4039b7e9c6a063e057fc : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d22db1d4aa9c4137a4f242ee3e6ffe6d : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd16c5f4456542ddb3d9e8aacfe5e169 : SUCCESS in 4m 05s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/9a3d66a03f524abca4e592915b2972c0 : SUCCESS in 39m 55s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/7b9cbc985c2e4915a7b4968a8e70c3c9 : FAILURE in 35m 39s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/390d916aa9204c96b05c3dc847c6863a : SUCCESS in 2m 10s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/ecbe011f13fa494da2597e3306509d92 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9e8aeb9e30c178504f072412b9e9682b752b27b8","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-12-07 11:23:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f619577167474d34ca865b30cded7aa024ed007b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 11:23:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d59b3ac24ad8f818f6b0782d2a262df2915aa0d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 12:07:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0bf144da8585406ab6aa146e15e9d356 : SUCCESS in 2m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba4e3530f5c240f191ff7b21db947df5 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84e6f11eff94d96b4c046b3b711ab2b : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0100a6b96e6c4a1d8b47b70d5119c3a8 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fd732aec56644eaad0f39487a2738fa : SUCCESS in 4m 06s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/b02a8147bb7a4ebe91efa1c7766b1233 : SUCCESS in 35m 57s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/8ff638112a3c4b39859a6b41888c1801 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"42abaa133e30fbab37669d3799ee489b63890083","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 12:07:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"429f515ba75d30b94866d44c073341af434c2ce8","revisions":{"429f515ba75d30b94866d44c073341af434c2ce8":{"kind":"REWORK","_number":1,"created":"2020-11-23 02:12:53.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/07/763707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/07/763707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/07/763707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/07/763707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/07/763707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/07/763707/1"}}},"commit":{"parents":[{"commit":"fd7c3bf7073b7afec3408de3c1d32ff137e82e37","subject":"bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/fd7c3bf7073b7afec3408de3c1d32ff137e82e37"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-11-23 02:11:39.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-11-23 02:12:42.000000000","tz":480},"subject":"Add py38 in setup metadata","message":"Add py38 in setup metadata\n\nChange-Id: Id2a9678b311af32bc6ec9a9832ce423042868e35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/429f515ba75d30b94866d44c073341af434c2ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/429f515ba75d30b94866d44c073341af434c2ce8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
