)]}'
{"id":"openstack%2Fmonasca-common~777203","triplet_id":"openstack%2Fmonasca-common~master~I41f47e02aef73bd9ebf6d6a46e8c212b9a268d73","project":"openstack/monasca-common","branch":"master","topic":"cryptography","hashtags":[],"change_id":"I41f47e02aef73bd9ebf6d6a46e8c212b9a268d73","subject":"Use C compiler to build cryptography","status":"MERGED","created":"2021-02-23 19:48:10.000000000","updated":"2021-02-24 10:07:19.000000000","submitted":"2021-02-24 10:07:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777203-cryptography","meta_rev_id":"82d3503461a0adbb1e850850d98a4f7cad16fedb","_number":777203,"virtual_id_number":777203,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 10:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-24 09:22:54.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-24 09:22:54.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-23 21:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 21:50:02.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"}],"messages":[{"id":"45c5e75fbccb541c0d16098df5dd6f7550487fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-23 19:48:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd990787de7ae7634459a90e37d451d9d525c29e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-23 19:54:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f0f19cb48b29dd142072d6fba973105ab7a401d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 21:03:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2d14eed2f0541a09cd732d48eef067c : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8d509bbc43941ed953d16849b506172 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1888b7a9a1c748c3a221c4002ae6e4f4 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5910d435966e41d0a6e4ed2d65461133 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4da3ce346f6c46b2b15cdfe95db967a0 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82ef09281e45428d83f3a6fde1290b64 : SUCCESS in 5m 39s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/50e30a872f564da492a4c3eb212c6621 : SUCCESS in 39m 39s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/d6e53df49aa648e8b7dc7ae91aa3272a : FAILURE in 33m 10s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/1bc6ecfbabb44070ac805a1c5fff3535 : SUCCESS in 3m 23s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/67d34ab5ae754a1288e7b1a4b2b7ebf4 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"73049d9804c06bf59070b5bc9026efa09ba746b7","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-02-24 09:22:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a983112da321307ab21e4f06c3d043d3dab337b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:23:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"401add81dc72a08ecbb93fec3cfd3832610cde33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 10:07:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8c59fe898144778ae165c09a8ee42b1 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8fc1b84f8254611ac4e6a45b1ba78fb : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ee38afa01a6447f9a5b5617e7e88b10 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9722586b0fc4f46938bb3cb05f4f2d4 : SUCCESS in 5m 17s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/18530a6c286b433cb7aa58cef178bafb : SUCCESS in 37m 05s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/dc8e95a1df72400e8262f15bc1a93287 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"82d3503461a0adbb1e850850d98a4f7cad16fedb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 10:07:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ccae320767e386b9445b99e2214b1ad23e1aa6d2","revisions":{"edb1c333a303583e85ebb380cdc9bcb8de580ddc":{"kind":"REWORK","_number":1,"created":"2021-02-23 19:48:10.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/03/777203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/03/777203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/03/777203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/03/777203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/03/777203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/03/777203/1"}}},"commit":{"parents":[{"commit":"f59a2f296ea9d2ff255de75ff39f7c807b755481","subject":"Bump librdkafka to 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/f59a2f296ea9d2ff255de75ff39f7c807b755481"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-23 19:40:01.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-23 19:40:01.000000000","tz":60},"subject":"Use C compiler to build cryptography instead of Rust","message":"Use C compiler to build cryptography instead of Rust\n\nBuilding wheel from cryptography\u003e\u003d3.4 uses by default rust compiler\nbut version 3.4.x keeps compatibility with c compiler by setting\nthe env variable CRYPTOGRAPHY_DONT_BUILD_RUST\u003d1 [1]\n\n[1] https://mail.python.org/pipermail/cryptography-dev/2020-December/000998.html\n\nChange-Id: I41f47e02aef73bd9ebf6d6a46e8c212b9a268d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/edb1c333a303583e85ebb380cdc9bcb8de580ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/edb1c333a303583e85ebb380cdc9bcb8de580ddc"}]},"branch":"refs/heads/master"},"ccae320767e386b9445b99e2214b1ad23e1aa6d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-23 19:54:45.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/03/777203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/03/777203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/03/777203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/03/777203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/03/777203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/03/777203/2"}}},"commit":{"parents":[{"commit":"f59a2f296ea9d2ff255de75ff39f7c807b755481","subject":"Bump librdkafka to 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/f59a2f296ea9d2ff255de75ff39f7c807b755481"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-23 19:40:01.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-23 19:53:57.000000000","tz":60},"subject":"Use C compiler to build cryptography","message":"Use C compiler to build cryptography\n\nBuilding wheel from cryptography\u003e\u003d3.4 uses by default rust compiler\nbut version 3.4.x keeps compatibility with C compiler by setting\nthe env variable CRYPTOGRAPHY_DONT_BUILD_RUST\u003d1 [1]\n\n[1] https://mail.python.org/pipermail/cryptography-dev/2020-December/000998.html\n\nChange-Id: I41f47e02aef73bd9ebf6d6a46e8c212b9a268d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ccae320767e386b9445b99e2214b1ad23e1aa6d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ccae320767e386b9445b99e2214b1ad23e1aa6d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
