)]}'
{"id":"openstack%2Fmonasca-common~777355","triplet_id":"openstack%2Fmonasca-common~master~Ie4b135c456333cd8a98cfb1d5231351786d1dd86","project":"openstack/monasca-common","branch":"master","topic":"fix-zuul-publish","hashtags":[],"change_id":"Ie4b135c456333cd8a98cfb1d5231351786d1dd86","subject":"Fix zuul publish docker image job","status":"MERGED","created":"2021-02-24 10:49:49.000000000","updated":"2021-02-24 16:31:29.000000000","submitted":"2021-02-24 16:31:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777355-fix-zuul-publish","meta_rev_id":"8ea8667bd0cb15a4cd1df6a76d67900dad026481","_number":777355,"virtual_id_number":777355,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 16:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-24 12:08:03.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-24 12:08:03.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 11:34:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"addc850f2ff24941e14d8d97ad36b595a1e2d7f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-24 10:49:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30572d9475d8c0189aa299b78daf4d2e2b84e784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 11:34:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6015e87e71b4e519cdf42e3fbc51cb0 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c46724db88044505bbfa8c3e67f270e0 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc6f729c549b46df9a5a829a14a85631 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a800c548d59047f982a6bb15ff5b0494 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6eb1b634855f48fca309c7a5bea0b9e2 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7991e3e966045048de92cad654c121b : SUCCESS in 4m 18s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/46328eca2bfd495ab43e06e51b653a05 : SUCCESS in 34m 17s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/eceb639052fa4304a3eb6f38334522bd : FAILURE in 29m 27s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/2a5b128cf7dc4cb9a415361c16643eba : SUCCESS in 2m 51s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/d649eca95a4644a2b65c8b0e1d79dddc : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f72505ed6071c0bf19c362508663c15b97d50ca2","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-24 12:08:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b45a03b68e2d2c1b40df11e23243d44838c72cbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:08:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0972f9eb516ad04f5c62b10b25686833e006e005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:52:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c7bf40f4dc34413b2f702d1fb917ba3 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c115d44f614ce7ace78ecb6254f2a4 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81861c955a534167b516985001ccfdd7 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/142f5bec8bae414b913939875e80d99e : SUCCESS in 4m 39s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/6be64e5f012b4df0b645fedf10bd5171 : FAILURE in 40m 50s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/fd044b9c899e4be9a324c0db0f7d6fd0 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"564804234022cc08e541e3680189176ca826e606","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-02-24 15:01:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1dfdfe53123772008cc90f0f4635d871db39eee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 15:50:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2ed99fad0946eaaf0a90e30d55275c : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9155f53e7d144347be361fcdb61bb078 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02dc182192864343b46e50ac5008df2b : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfab903082c944ec90835a33d321e4cd : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/961b62144df943bd93bc248f6c57f9c0 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e8e98e077c6414880336f6581230d8d : SUCCESS in 5m 04s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/4dfdc22e065c4905a5c6d7b4e4301316 : SUCCESS in 41m 45s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/939cda0c1c9948faac0e36dc0772324a : FAILURE in 31m 59s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/8b05cc9185584837990a46696cacfae7 : SUCCESS in 2m 45s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/69bec8801d6940f994cd5a4001436efa : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"47f9f3d21fb36153554e941758be06a5faf08e2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 15:52:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36ea95258537c2fd8d2609d13bda7b88209aa8ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:31:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e53e939ad784bceb27333951fdcc1d7 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7493a97711944b5aebc94d5a3326b9e : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2516f37174444d0a06869c1e1c4e081 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84c310083b694db58b92a29d6e73e5cd : SUCCESS in 4m 35s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/cc55a8ae415e40959db356bb73cb9bde : SUCCESS in 33m 08s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/305597db7988413c85591026f4ea1177 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8ea8667bd0cb15a4cd1df6a76d67900dad026481","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:31:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9aa47afe5d39efcfd2f8b64810ac58857b4f7997","revisions":{"9aa47afe5d39efcfd2f8b64810ac58857b4f7997":{"kind":"REWORK","_number":1,"created":"2021-02-24 10:49:49.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/55/777355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/55/777355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/55/777355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/55/777355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/55/777355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/55/777355/1"}}},"commit":{"parents":[{"commit":"ccae320767e386b9445b99e2214b1ad23e1aa6d2","subject":"Use C compiler to build cryptography","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ccae320767e386b9445b99e2214b1ad23e1aa6d2"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-24 10:48:20.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-02-24 10:48:20.000000000","tz":60},"subject":"Fix zuul publish docker image job","message":"Fix zuul publish docker image job\n\nAdd tag to docker push command with specific branch name\n\nChange-Id: Ie4b135c456333cd8a98cfb1d5231351786d1dd86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/9aa47afe5d39efcfd2f8b64810ac58857b4f7997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/9aa47afe5d39efcfd2f8b64810ac58857b4f7997"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
