)]}'
{"id":"openstack%2Fmonasca-common~795119","triplet_id":"openstack%2Fmonasca-common~master~I43e86ff631034bc6b35fedbc30b72282b5d3b7c8","project":"openstack/monasca-common","branch":"master","topic":"docker-build-fixes","hashtags":[],"change_id":"I43e86ff631034bc6b35fedbc30b72282b5d3b7c8","subject":"Fix Docker build","status":"MERGED","created":"2021-06-07 14:45:50.000000000","updated":"2021-06-07 16:37:51.000000000","submitted":"2021-06-07 16:37:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795119-docker-build-fixes","meta_rev_id":"bcefff44649cb20f1f3cab30146f1e009cee53af","_number":795119,"virtual_id_number":795119,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 16:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-07 15:59:17.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-07 15:59:17.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 15:36:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2acf305b8b9cea99dc4ff677491079996149d2fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-07 14:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e72426c9afa427a8250892516bcf3a017e941b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:36:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc67aed68e224f73a6e84ce11e57cd30 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/949d57d888f94f7291833f5696cf60e9 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f375a3ef3d84c488d0fe53c4e7c9b2b : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f8b5296186c40ada51feb90e9e94482 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a7968876c2342cdb911016c3caaa0fa : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/189d3a89afea403482c45551ae491b79 : SUCCESS in 6m 05s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ec61dc2d410f448eb1f375f30216622a : SUCCESS in 42m 28s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/5c077104eabb4a008ce57da5cfe10c60 : FAILURE in 30m 08s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/eb45bb227bff47a382109c1713972404 : SUCCESS in 3m 26s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/6e3c8b049ad6408296e2562c6f1f20ad : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c7ef7eb78e7aaf58df92cf70ff130ae89734a1ff","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-07 15:59:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd9a7a059fa2aa94e62892566a7f4a0fb3d2f615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:59:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ed4ffe063bdcdea0db40e6b3ce91657976d0f67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:37:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82322d414e374305a0b392464b4dce64 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d53cdc2748473ca442484ceb2473a6 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85cde063dd2d43399b5a3017c1cade8b : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31e47185faa0481eaea303a710f2780d : SUCCESS in 5m 02s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a0abfae579de41df87af38d4e2e27086 : SUCCESS in 36m 37s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/36d0840fb63843aba2320f5ab63634d5 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bcefff44649cb20f1f3cab30146f1e009cee53af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:37:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"844db67a91fabfcf7ca8ec55aeb3a700ab1b6168","revisions":{"844db67a91fabfcf7ca8ec55aeb3a700ab1b6168":{"kind":"REWORK","_number":1,"created":"2021-06-07 14:45:50.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/19/795119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/19/795119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/19/795119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/19/795119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/19/795119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/19/795119/1"}}},"commit":{"parents":[{"commit":"db200fd98b8eaca5e76f4284570eed0c4169c892","subject":"Bump sublibraries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/db200fd98b8eaca5e76f4284570eed0c4169c892"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-07 14:44:07.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-07 14:45:01.000000000","tz":120},"subject":"Fix Docker build","message":"Fix Docker build\n\nThe default nodeset was just changed to focal [1], which in the absence\nof the python-is-python3 package doesn\u0027t provide a python binary.\n\n[1] https://review.opendev.org/c/opendev/base-jobs/+/789098\n\nChange-Id: I43e86ff631034bc6b35fedbc30b72282b5d3b7c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/844db67a91fabfcf7ca8ec55aeb3a700ab1b6168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/844db67a91fabfcf7ca8ec55aeb3a700ab1b6168"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
