)]}'
{"id":"openstack%2Fmonasca-common~826343","triplet_id":"openstack%2Fmonasca-common~master~Iacb7d3b5eeb1540647c189d76cf5aa3dee9706b5","project":"openstack/monasca-common","branch":"master","attention_set":{},"removed_from_attention_set":{"16957":{"account":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"last_update":"2022-01-25 22:03:05.000000000","reason":"Change was submitted"},"26141":{"account":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"last_update":"2022-01-25 21:30:15.000000000","reason":"Adrian Czarnecki replied on the change"},"28062":{"account":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"last_update":"2022-01-25 22:03:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacb7d3b5eeb1540647c189d76cf5aa3dee9706b5","subject":"Update Monasca Base Image","status":"MERGED","created":"2022-01-25 17:36:29.000000000","updated":"2022-01-25 22:03:05.000000000","submitted":"2022-01-25 22:03:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826343","meta_rev_id":"7c6a20950b19b9d6e146edf3d095b71dd48be377","_number":826343,"virtual_id_number":826343,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 22:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-25 21:30:15.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-25 21:30:15.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"}],"reviewers":{"REVIEWER":[{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 17:37:42.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2022-01-25 17:43:01.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"state":"REVIEWER"},{"updated":"2022-01-25 18:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b675be596416f0d6c4cafc02bdf708b6683ead4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2022-01-25 17:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52552c7f6729ebb633fffcf77666a8f686c21df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:32:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ee4a4e1b08a4f7ca0c1aa2e52c04ec2 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a5d90b8ac7f404eb183c31a738eb8fa : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b01dc548bef48f794d85cf23b9f8d5a : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d3ddfa807c040138cfe369598870a21 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68ac496c6ad44e91b2829fefdd50d92a : SUCCESS in 4m 47s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a11bfef4b9e54b8fa78b776b5611e7fc : SUCCESS in 46m 13s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/db162adf4fd84ff49e23eac7fa1f2030 : FAILURE in 41m 37s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/12bc053d54da4ba6a5d30103d89cbb6e : SUCCESS in 2m 54s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/4a721dded1794e18a64f1d5ad0165f4a : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5332b7903e0868a439ae941189271d9495851862","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2022-01-25 21:30:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"665184fc72bb4e10016b72f97dbb336f3f1c8953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 21:31:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a784d03bf2acd3b6dffab3a7affc977fc0fe495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 22:03:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42776813fe7a4d098dc6549255b4e3ab : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a290ab79cd9340019b4505a98d994030 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf93506fd946432b9f11cd97d9dc5252 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d5f7c4f02354d04996b84e1ff14387a : SUCCESS in 4m 41s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/db72b09150a24953a451c9d70cbdef66 : SUCCESS in 31m 35s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/82bb03a0ad654ea89ff4eaaf382e1193 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7c6a20950b19b9d6e146edf3d095b71dd48be377","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 22:03:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffce75af6c8413ec9952c3ebfceb6d99493c83b3","revisions":{"ffce75af6c8413ec9952c3ebfceb6d99493c83b3":{"kind":"REWORK","_number":1,"created":"2022-01-25 17:36:29.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/43/826343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/43/826343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/43/826343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/43/826343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/43/826343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/43/826343/1"}}},"commit":{"parents":[{"commit":"d6ebd62a6daa55a81368d2e70ded9843fa08ff54","subject":"Bump librdkafka dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d6ebd62a6daa55a81368d2e70ded9843fa08ff54"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2022-01-25 17:30:14.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2022-01-25 17:30:14.000000000","tz":60},"subject":"Update Monasca Base Image","message":"Update Monasca Base Image\n\n- Bump base image to python:3.8-alpine3.15\n- Add Rust compiler and its dependency Cairo\n- Bump librdkafka to v1.8.0\n- Removed env variable CRYPTOGRAPHY_DONT_BUILD_RUST\u003d1 since\n  new Cryptography requires Rust.\n\nChange-Id: Iacb7d3b5eeb1540647c189d76cf5aa3dee9706b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ffce75af6c8413ec9952c3ebfceb6d99493c83b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ffce75af6c8413ec9952c3ebfceb6d99493c83b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
