)]}'
{"id":"openstack%2Fmonasca-common~855735","triplet_id":"openstack%2Fmonasca-common~master~Idb97bb8f2cabc59126958205e137d8e70e11b1b6","project":"openstack/monasca-common","branch":"master","topic":"release-health-check","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-07 09:29:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idb97bb8f2cabc59126958205e137d8e70e11b1b6","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:30:52.000000000","updated":"2022-09-07 09:29:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"65b18a9d30683475570bd0e6542f5810d2b71696","_number":855735,"virtual_id_number":855735,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-03 21:36:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 21:36:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7933733fe1e0c9193727db80a27f4073e5311ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:30:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d1cefca281290a7682fb525ccf127e920636351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 21:36:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc2613479370495ca1ccaec148fc9428\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21a4e24e3922499f9b8ec652ca750ed5 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b43117ee8e4baea8f9347318286fc3 : FAILURE in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b677ec86bbd743d6a6b36f086daae782 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f00940e13aa7461699a0e5c8ba46cdcf : SUCCESS in 10m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d24c198f19f849d684ed950858f22487 : SUCCESS in 9m 44s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e09b17f418734e5a86c8ba55d5f66e89 : SUCCESS in 53m 02s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/4a6c3f9bfde24a94988ad3538655f84e : FAILURE in 46m 05s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/4a63d998bd924405a04f778c0a125e1b : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"65b18a9d30683475570bd0e6542f5810d2b71696","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:29:00.000000000","message":"Abandoned\n\nGate check passed (pep8 job failure is related to the health check patch), no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26828bc4667b6dda4e8352e94a1ae928f28757a1","revisions":{"26828bc4667b6dda4e8352e94a1ae928f28757a1":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:30:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/35/855735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/35/855735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/35/855735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/35/855735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/35/855735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/35/855735/1"}}},"commit":{"parents":[{"commit":"f290882658db757a4dfbde23c327b70cbe58f964","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/f290882658db757a4dfbde23c327b70cbe58f964"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:30:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:30:50.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Idb97bb8f2cabc59126958205e137d8e70e11b1b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/26828bc4667b6dda4e8352e94a1ae928f28757a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/26828bc4667b6dda4e8352e94a1ae928f28757a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
