)]}'
{"id":"openstack%2Fmonasca-common~874297","triplet_id":"openstack%2Fmonasca-common~master~Id9c0956d4800f01dadfece02638bc4fcc9ef1e8a","project":"openstack/monasca-common","branch":"master","topic":"fix-zuul-errors","attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-06-15 06:53:22.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Id9c0956d4800f01dadfece02638bc4fcc9ef1e8a","subject":"Fix deprecated zuul queue syntax","status":"MERGED","created":"2023-02-17 19:20:25.000000000","updated":"2023-06-15 06:53:22.000000000","submitted":"2023-06-15 06:53:04.000000000","submitter":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874297-fix-zuul-errors","meta_rev_id":"3d0219efeb02220c2e342ff9f3a66e8f35960913","_number":874297,"virtual_id_number":874297,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-15 06:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-17 20:21:31.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-15 06:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-15 06:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 20:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-15 06:52:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a5335795df576c90874b6e2ab3d2271e2eb85925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-17 19:20:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34c61d29b9c04b4ec81911a3bfe877ae4a07847b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:21:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b22577b8cadf42dbb2b3af89688cd8ea\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/86d185895efe43a680512a3340a965aa : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8b5b735b0624eab98bbcd4afe033617 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96834d601b8d4d8893cc149267ab9373 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f95e7c39a91d40b6b322744978f982dd : FAILURE in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3e2c48af5d346b4a680abb18136ee45 : FAILURE in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74aaa542e4a948f7961549d0fde4bc83 : FAILURE in 3m 53s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/42c4f8ab11d1441884a38acd4143f6ac : FAILURE in 37m 38s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/957b6e1f1ab141b6991631dc87bd1b65 : FAILURE in 33m 08s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/52714fc2523b48aebcc9b2a8d2f2e025 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fe729eb0753405db7d343c49bc12df0833f49b5c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 06:48:46.000000000","message":"Topic changed from fix-zuul-queue-syntax to fix-zuul-errors","accounts_in_message":[],"_revision_number":1},{"id":"8ea3a279602901cff7f5e7997abff68f37fa67ff","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 06:52:53.000000000","message":"Patch Set 1: Verified+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c40d70122895686a5f23e5378c9826e9678c0c11","tag":"autogenerated:gerrit:merged","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 06:53:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3d0219efeb02220c2e342ff9f3a66e8f35960913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 06:53:22.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/monasca-common on branch master.  The error was:\n\n  Job monasca-tempest-java-cassandra not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      queue: monasca\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-python3-zed-jobs\n      check:\n        jobs:\n          - monasca-tempest-python3-influxdb:\n              voting: false\n          - monasca-tempest-python3-cassandra\n          - monasca-tempest-java-cassandra\n          - monascalog-tempest\n          - legacy-monasca-common-maven-build\n      gate:\n        jobs:\n          - monasca-tempest-python3-influxdb:\n              voting: false\n          - monasca-tempest-java-cassandra\n          - monascalog-tempest\n          - legacy-monasca-common-maven-build\n      post:\n        jobs:\n          - legacy-monasca-common-localrepo-upload\n\n  in \"openstack/monasca-common/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ca41523731c0f78e7322d6770a9dcdc1710b7ba","revisions":{"1ca41523731c0f78e7322d6770a9dcdc1710b7ba":{"kind":"REWORK","_number":1,"created":"2023-02-17 19:20:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/97/874297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/874297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/874297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/874297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/874297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/874297/1"}}},"commit":{"parents":[{"commit":"f290882658db757a4dfbde23c327b70cbe58f964","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/f290882658db757a4dfbde23c327b70cbe58f964"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-17 19:18:55.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-17 19:20:16.000000000","tz":60},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nmonasca-common.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n\nChange-Id: Id9c0956d4800f01dadfece02638bc4fcc9ef1e8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/1ca41523731c0f78e7322d6770a9dcdc1710b7ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/1ca41523731c0f78e7322d6770a9dcdc1710b7ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
