)]}'
{"id":"openstack%2Fmonasca-common~907497","triplet_id":"openstack%2Fmonasca-common~master~Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d","project":"openstack/monasca-common","branch":"master","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2024-02-08 05:48:45.000000000","reason":"\u003cGERRIT_ACCOUNT_20865\u003e replied on the change","reason_account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-02-22 12:09:54.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-22 12:09:54.000000000","reason":"Change was submitted"},"35988":{"account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"last_update":"2024-02-22 11:30:18.000000000","reason":"\u003cGERRIT_ACCOUNT_35988\u003e replied on the change","reason_account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}},"7556":{"account":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"last_update":"2024-02-01 21:17:58.000000000","reason":"\u003cGERRIT_ACCOUNT_35988\u003e replied on the change","reason_account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-22 12:09:54.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-02-22 12:09:54.000000000","reason":"Change was submitted"},"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2024-02-22 12:09:54.000000000","reason":"Change was submitted"},"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2024-02-22 12:09:54.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-02-22 12:09:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d","subject":"fix: Upgrade librdkafka","status":"MERGED","created":"2024-02-01 20:38:45.000000000","updated":"2024-02-22 12:09:54.000000000","submitted":"2024-02-22 12:09:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"907497","meta_rev_id":"d795cd070be9a4edf77e099cb5f7c9ec98580b8d","_number":907497,"virtual_id_number":907497,"owner":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 12:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":2,"date":"2024-02-22 11:27:49.000000000","_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":1,"date":"2024-02-22 11:30:18.000000000","_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}],"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 21:17:31.000000000","updated_by":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"reviewer":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"state":"REVIEWER"},{"updated":"2024-02-01 21:17:58.000000000","updated_by":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"reviewer":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"state":"REMOVED"},{"updated":"2024-02-01 21:17:58.000000000","updated_by":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2024-02-01 21:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 13:30:02.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"CC"},{"updated":"2024-02-05 13:30:02.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-02-05 13:31:10.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"},{"updated":"2024-02-06 17:09:09.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2024-02-07 06:27:46.000000000","updated_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"reviewer":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"state":"REVIEWER"},{"updated":"2024-02-07 08:18:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-02-08 05:48:45.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-02-08 05:48:45.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-08 05:48:45.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"08304e82b91f6f4fa079387e861968b477cfcec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-01 20:38:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cb442115d8eb33feb9349ed52a89654c0a6b8ca","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-01 21:17:58.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_7556\u003e.","accounts_in_message":[{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"}],"_revision_number":1},{"id":"43bdd5675b2e43b6e75ce83528e3f7e1df5cb91a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 21:26:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a5087af1be64c0c8564ffd7879341f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3c6888994004a0eb6d0f1fc329ffdd3 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/683ca0a29acd4ebeb54639743cda696c : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7b5ca6d7c7a4a4f90f21575cc9e9d41 : FAILURE in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7666ae9cddd4318a4411c808bd7d526 : FAILURE in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39ac10ef69534916afb5dc44c928b0d9 : FAILURE in 2m 46s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f510ee54be4445d8bcdc7c9cdaedd811 : FAILURE in 37m 54s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/ce350f07d7f04ea3bfbf9909dd9d0e87 : FAILURE in 16m 16s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/c11dd669e45c4596bfdbce57ae38201d : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"252906760114fcad5d05e91c602d7016a641b4f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-01 21:40:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bf50886330498fe1c4adfcbc2eeedd3d2cee2404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 22:35:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd1ab814cea24c4b9d590b0506b3e1eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6402509070d44a69a0bb19d694c1181d : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52bd440d0e14d0d9cde1fee70dbcb68 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5b90849bd8c4458823d5961b60a28b8 : FAILURE in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d3737d30cff4a8281945dafb577013e : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7410200635984ae1a7d55dfe6f61a8ee : SUCCESS in 4m 34s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/76fab6100a9e429e88ec8f54b6617f94 : FAILURE in 45m 54s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/1700fac7581042779db50cdd2bf58559 : FAILURE in 23m 43s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/686b03833443493a805f2547a2baf888 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d2b0f828d579b6ce3312a8bc3bdbfab3c56cc06c","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-02 05:52:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d179d81734a94faca8c064e87a4058083b110f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-02 06:23:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ef46b299a6e26cb26c151b293d09ee4344e54162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 07:01:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c723fb850a649c887bc6abc7112ca69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0202f8603d304fa1affc371eae909615 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1f1069244c45fa8f074f6bc9dc5e27 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d04d5d910fa40a994e5d706bebe1d11 : FAILURE in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f59a9ae780d849f39777af575ffe95cd : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea76a11ce32e441e96b908f3ae4ad79a : SUCCESS in 3m 12s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/22a31f1dcc534e3bbed9a7a334a2c915 : FAILURE in 35m 31s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/122929c30cc842098906cd645ff507eb : FAILURE in 24m 51s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/9abbc7d3bdc6469b8c3d64ccb9252c11 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"99c73b9a73de47787f4a215ae826296216098326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-02 07:58:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"365c81a5692bb475b5d3447b8ba83030d2a1be80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 08:55:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb8dca99703542b68355950bae6465cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdc4501c9e044ea98d872ec3efc7945d : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bab6b6f5a24e9f8448f08894f71662 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c1984f6bc2243b08aa1c1a0dc86093e : FAILURE in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/993b2c55547047ad8ac8023a44ca250b : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bed0655e87274d0194df13839bf6cb63 : SUCCESS in 3m 31s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/6fc7b756bea843689bbbab7584a0608f : FAILURE in 50m 32s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/3f8a9e284e7d4413bb184891f3a513d4 : FAILURE in 25m 06s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/1918a72daabc4652a6533d74c51f191a : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c76afb26612ccc6a486915a5adac7530876487d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-02 11:42:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9c780cedf87f3e0db49193369dc29f82f4bb0a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 12:13:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51ebd2821d104747b1f6c3444b95fa3a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c87e57df8c5b4e99b058403a7298753d : SUCCESS in 2m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ffd8c76d7ef4c47b5eff54da6c4dd19 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfcc9c742ef4af1a27928f9b108b106 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97f6cde7d39b4e7dad4d1dcad74aabf7 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/136766681f7a4850b1feea559eebecf0 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a437ec894cb74336b2f4eb53513e4fa9 : FAILURE in 3m 30s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/249b00e661ff445cbfb466e06b289f21 : FAILURE in 29m 35s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/9b700503c7ba4a5eb3738dfe5ca30221 : FAILURE in 13m 07s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/72a9e21af11d42d5b65d460d9eae478e : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":5},{"id":"83371ade86aad7117d3fbd6ae2af82ffa5b53066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-02 12:37:06.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ff5136bea7f75f9a34c288583497203c6b935a6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 13:29:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab4430337ae47e3937390c977046b47\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/276f5a0d97e84d1f9197879fe37cf71a : SUCCESS in 1m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ce1b5a3b405488dba84580ea44af7ba : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e89b3811a749cf81710f9809cdfaff : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/641af94bbd4d4fe387fd46701aa2ef25 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31ddf7ee6bc34eaa9745ed8bc3d984d5 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d7f9baa1277495f883bd7fa1e7ad630 : FAILURE in 3m 19s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/d5102c44a6a54c948d66203b31977916 : FAILURE in 46m 46s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/92ef47a1614b4365a8483e3d8c0e89d8 : FAILURE in 13m 04s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/55fbaff30d5349e2bf30f1f4d8cfc3de : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":6},{"id":"96ea26072803a1e17751101fcc00f6cb92d509c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-02 13:36:18.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"88f34a74337891676b75f9c548105cdcaa05ed61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 14:27:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fd6d85cc5e44f5ea19ad7fe3e72b834\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e8e06fb370ed4cdda1a5633442ea39b4 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b084bd07dea1488ca32f082b0620b861 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01cd5d2e17f94bb59e3c3393c1bd1465 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e22a7f6c4342beade96c79b8bd778f : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee69b5b4862144c3ace706282b48d73b : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef300468c3914b95a8c355de95804fef : FAILURE in 3m 30s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/aa2387d5878a49feb71a82d6e216a583 : FAILURE in 44m 51s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/a6da1dac07c74ed3bad13e3a09657a92 : FAILURE in 25m 17s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/742795104e1844918ac634dfed0a18ef : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":7},{"id":"c9eff72328ccbb2ec3f0c400fd82ff9ea41115cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-05 07:32:37.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"4cdb49936adf4796396cee2854a381756752a897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-05 07:42:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e70070296109e9a5d8dea39bf4353492d781d098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 08:21:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60a237dfab384e779d3c22eb9b07823f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dfbd9b1ed8e644198dd808969fe77e36 : SUCCESS in 1m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9346dbdead64ca8820a4b1166af5a4e : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dababb90dc8c4cbd888793254c225b41 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c80af9215c54cd99c6407e11b8f2e56 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3537c2e7a239470983e419f7967f2204 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af23b8787ea341e2aed3d7b69683c5d7 : FAILURE in 3m 13s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f6991a8910614ff9b3a14a74b00fd7e5 : FAILURE in 30m 55s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/becaa98fca184e63a02605dfcdcc4b1d : FAILURE in 28m 09s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/9271c313db364fa6ac27a176fa30e72d : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":9},{"id":"b732a286c263d9ab12005497479981daa17e1fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-05 12:08:30.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"da904ccf36c472bafd6d602eeab8ca557a8b5d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 12:41:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c02463206184dde830cd1a6e4a5c4f9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21fef3f6bc1f4914afa11adeb5961899 : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f4894d531af4034bfe74bd453f9e17e : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1ea31a307e4076a5628d50544e92af : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c35e9f5ae7f54a598db2bbc4dd024545 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fe55b7d3b4c4088b9bb823332e65769 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe7be634c23241c8a0dba62ad4171340 : FAILURE in 3m 24s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/8f022003cadf40ebb48bc5f48707dbaf : FAILURE in 31m 34s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/fe9f20f616ea45d3a83435ec98799533 : FAILURE in 22m 43s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/fe640a6829434d81b2c4e1ba3ce98e4b : SUCCESS in 4m 14s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/25f20d95b40c46cabe4d9ebf96dabb71 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":10},{"id":"e20c2df8f06b2a62c915c7c9dec304926902a717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-05 12:51:32.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"ee3fbd204add52850690341585bc9209f4b4d43b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 12:52:08.000000000","message":"Patch Set 11: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job publish-monasca-common-docker-base-image","accounts_in_message":[],"_revision_number":11},{"id":"c545501d87f2937300c104d31ceeb6014d22c98c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-05 12:58:33.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"3ece540d02925be949f2c48546a72a06a4a2251f","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-02-05 13:30:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e8a1f5ec3acab1b8ddbdec1d3c19c71c9c83d9b7","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-02-05 13:31:10.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"450fabf225180bda16271a219c82bc531840db48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 13:47:33.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccab6fcf57954149817d83a38e5889ae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8378aa7b5bdd41b6bb8e3d684e1902df : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13e9c210501a4c32b9f8e375b6090ba4 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b7d12368e34ffb8a728968e7e35e26 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48ee2930d06744659d865688704ae28d : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61d759632b78411bbf786332aa71aaa1 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f426d2d0bf54058a1e536b9cccd6e32 : FAILURE in 3m 21s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e2478fd300834b91a30f0ccb759d363c : FAILURE in 41m 43s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/4b7e1700fad149daaf37fbcb8a806528 : FAILURE in 43m 12s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/1ff92f5ae9a0483ca47a6c92789a4e5f : SUCCESS in 3m 47s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/cd75c435cc1f43738bab7165d961ae16 : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":12},{"id":"0f9339c4b2de266c9da75d23132f1629da59d114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-06 08:44:02.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"4066a15fd8322fbeff907a70696445ec70115306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 08:44:39.000000000","message":"Patch Set 13: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job publish-monasca-common-docker-base-image","accounts_in_message":[],"_revision_number":13},{"id":"1d9569d45f989cb4ac843dff08a9dee62fbd2360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-06 08:49:15.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"ead049d09aa2fd29ef790a847fe5bc500ad3c122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 09:55:21.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476ef19d3c854607afcf5f552712d9bb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c630f114b10d476ebe5bb477520e2a26 : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12984ad85cf442b0a3ab849718d4cb0d : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdf186a32df46a2875f1de6d61e6d54 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e4b1357a0034c5b889b064cef1ba53a : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecdd86d94ccd4b8aad723ab84f078a0e : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9c8ef374ac744c0be9563bc03e10555 : FAILURE in 4m 01s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/03acb854a3244c069c44a18cede5196c : FAILURE in 39m 27s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/56fa850d25164341b890eb3494a60751 : FAILURE in 25m 29s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/6a7f701ea4be4768b06a5e8db06b488e : SUCCESS in 3m 32s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/7af883e50ca64d6599c1df592cb62a48 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":14},{"id":"d39632caab3e88c5b793e2211d3c61f8763fe99d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-06 12:53:50.000000000","message":"Patch Set 15: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"ee4f17359670cfd52fd96c588a9d3c70e19b7f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 13:24:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1231ee7c3410461e80ac8ba77ad1a521\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/90709757d8e647f8a9b8b7c4d852495e : SUCCESS in 1m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e94ffbeeb3234e8f84fca14de705b281 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80a02e9aba104ec5ada37b49e08f4c0f : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50a9fc258a1b475da41b65ab154be94a : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a613f35c654403b9dbae3465f73de72 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91da704349f6411c8e8f8ef55e8fd23b : FAILURE in 4m 54s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a628a3383d7c45f5904396c251b39c6e : FAILURE in 29m 21s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/0a9232eb02e041a4ae64b174b6429a79 : FAILURE in 11m 55s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/a6cf0a17b5d447d5848d17c9329c8dc2 : SUCCESS in 3m 42s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/5e8d40e433254dabb5b20fd499ab363d : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":15},{"id":"76f9dd8cee8067c03ba40c5b91224565817441db","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-06 14:45:15.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"98038907eb0f93d916bb78abe719d32482294772","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-02-06 17:09:09.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5c67c92006138f473f235e837f35a6cda35ccc86","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-06 19:05:59.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cd376aa0b36dbcf336c52e15d9fb8610e5dedb50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-07 05:59:16.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"6fdd659e0ba3a0804291020d14488025d3dcf163","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-02-07 06:27:46.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f406a2d46d1c90048c37e91b362257141d0b36a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 06:53:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fabc948e192494aa6dfacdb29f0c2ba\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/06a4c8f6e03a442787d7a9d760d527e2 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffbb98d6cc5f4d01b6bd6a2ac212678a : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec384b6fcd8f468b98616a0e2938a06c : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eec7fe6df3aa4000b2a895c42869cd8b : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/829aa1192e6c4c8996e1777cc10f88c6 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a45d641dac7f42f4a7456087c2e5986d : FAILURE in 3m 07s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/6bc45b519a3b4e1db3f2a1e2d75080a6 : FAILURE in 47m 44s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/3d6f3464ec3247b29aee75c21ad1083e : FAILURE in 26m 08s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/11f2ac9e0be24623a46dd36f88e93201 : SUCCESS in 3m 23s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/3c5548973e0e42db81849bd1a2804165 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":16},{"id":"a0e6af75ffc14ba82357415df9312f34c3f1cb51","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-07 08:18:14.000000000","message":"Patch Set 16: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"a6e35b37e2d83176cb8de65e3b8e4b5cdd0cfb72","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-02-08 05:48:45.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"d477368852132f606c67a54796eedb60ff62feea","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-08 10:27:13.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b679ef4fc10cecaac4cf3358d9abbed60244a413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-08 16:11:29.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"cc0c152a09a7f37288fd045ad600e28e50f49beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 17:01:03.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18a3327bf994429c8a9a93577a2865da\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e999e4dc8db0450b9be7df6c530b9239 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc2c03e09492493a959b6412fe208e27 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5016a8bf924c1da69d017810b2955f : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c115b4d2aa934e97a3e4666e6a0f9f07 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/550d6ab09e1e42a9a7d4f41bab8c05ca : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf0863afb452484baecfac4ea4601c3e : SUCCESS in 3m 34s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ac19c8ece1b34bda87ca4d5746c81c5b : FAILURE in 33m 56s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/3b2fceb20e22422c9e87ff315de568a4 : FAILURE in 14m 47s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/d05b6b012f7d490085a2d2ed24d70c37 : SUCCESS in 4m 41s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/00a939cb6b3447f0ab422635bb426e8f : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":17},{"id":"92570105e61c4ec117fe66b0d058ee0bbdd347dd","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-08 20:48:18.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6bc2cc7e136a2027088f6aace94a1724b9b84653","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-08 20:55:29.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"0c20fac9f08d7bddf92d6cd5b299fe823dc3f557","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-02-09 07:03:15.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5ccc53760236af856a897e23201075c3d56f3ae8","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-02-22 11:27:49.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"bcab33ba1a4b7c2b6d27de42670648e127780608","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-22 11:30:18.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"9964e3222ddfca6111f5345d2b1865e56f91efa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:31:04.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"4f797ef84bdba98e1f94972e2359ebc3298f9682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 12:09:54.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd636b6641cd4d3f9a487c67a21aa0bf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cf91a14541b14e289aa2b9fd1f80dbb4 : SUCCESS in 2m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e46221b99504ec3a52d2217c1f0d3d1 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/948e70cc918c42b1a92e220cda09c9e2 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5c4da8395aa410a8a9a46e4043cbeb4 : SUCCESS in 3m 40s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/b46890dab84747c6bcabea6f814c7977 : FAILURE in 37m 49s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/84f54ec1b0df45b9924afa5422e759ad : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":17},{"id":"d795cd070be9a4edf77e099cb5f7c9ec98580b8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 12:09:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"fe601fa51b0e5887f9557dab0765b4e415b418eb","revisions":{"e351be76c08be34e36bb6bd5db3b0c8f59f849d4":{"kind":"REWORK","_number":1,"created":"2024-02-01 20:38:45.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/1"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:39:39.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e351be76c08be34e36bb6bd5db3b0c8f59f849d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e351be76c08be34e36bb6bd5db3b0c8f59f849d4"}]},"branch":"refs/heads/master"},"2f34308ba8bc58d54a151a64f25aec00f413c72b":{"kind":"REWORK","_number":2,"created":"2024-02-01 21:40:47.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/2"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 21:41:33.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2f34308ba8bc58d54a151a64f25aec00f413c72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2f34308ba8bc58d54a151a64f25aec00f413c72b"}]},"branch":"refs/heads/master"},"cb4c2b4c405116040df02094b040a57afd57bf06":{"kind":"REWORK","_number":3,"created":"2024-02-02 06:23:29.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/3"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-02 06:24:12.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cb4c2b4c405116040df02094b040a57afd57bf06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cb4c2b4c405116040df02094b040a57afd57bf06"}]},"branch":"refs/heads/master"},"8cc0daa2036a2f70d60af13e82a98cbc3e0f9dc9":{"kind":"REWORK","_number":4,"created":"2024-02-02 07:58:55.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/4"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-02 07:59:40.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/8cc0daa2036a2f70d60af13e82a98cbc3e0f9dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/8cc0daa2036a2f70d60af13e82a98cbc3e0f9dc9"}]},"branch":"refs/heads/master"},"45854dcf5de116d487867c038088ed1ab9b51cf5":{"kind":"REWORK","_number":5,"created":"2024-02-02 11:42:26.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/5"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-02 11:43:10.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/45854dcf5de116d487867c038088ed1ab9b51cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/45854dcf5de116d487867c038088ed1ab9b51cf5"}]},"branch":"refs/heads/master"},"cbf06af792b5636df3ee61ac0981439a6d97f348":{"kind":"REWORK","_number":6,"created":"2024-02-02 12:37:06.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/6"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-02 12:37:56.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cbf06af792b5636df3ee61ac0981439a6d97f348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cbf06af792b5636df3ee61ac0981439a6d97f348"}]},"branch":"refs/heads/master"},"d11965ba52f7cc54e32c1d82802abc3e96398b26":{"kind":"REWORK","_number":7,"created":"2024-02-02 13:36:18.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/7"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-02 13:37:06.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d11965ba52f7cc54e32c1d82802abc3e96398b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d11965ba52f7cc54e32c1d82802abc3e96398b26"}]},"branch":"refs/heads/master"},"77631d1e4f7b53ad6e712f10343bc5f73c16c07f":{"kind":"REWORK","_number":8,"created":"2024-02-05 07:32:37.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/8"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-05 07:32:56.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix : libressl3.4 packages removed\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/77631d1e4f7b53ad6e712f10343bc5f73c16c07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/77631d1e4f7b53ad6e712f10343bc5f73c16c07f"}]},"branch":"refs/heads/master"},"3fa80e6e67b56c5e505dd1aa701779baade751f9":{"kind":"REWORK","_number":9,"created":"2024-02-05 07:42:56.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/9"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-05 07:43:50.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix : libressl3.4 packages removed\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3fa80e6e67b56c5e505dd1aa701779baade751f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3fa80e6e67b56c5e505dd1aa701779baade751f9"}]},"branch":"refs/heads/master"},"f7122611f467d962a678a8d301ad96748918a9c9":{"kind":"REWORK","_number":10,"created":"2024-02-05 12:08:30.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/10"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-05 12:09:02.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix : libressl3.4 packages removed\n\nfix: publish docker image jobs added\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/f7122611f467d962a678a8d301ad96748918a9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/f7122611f467d962a678a8d301ad96748918a9c9"}]},"branch":"refs/heads/master"},"256f2c394fe459ef7174da9cb47606b0d516419c":{"kind":"REWORK","_number":11,"created":"2024-02-05 12:51:32.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/11"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-05 12:52:28.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix : libressl3.4 packages removed\n\nfix: publish docker image jobs added\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/256f2c394fe459ef7174da9cb47606b0d516419c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/256f2c394fe459ef7174da9cb47606b0d516419c"}]},"branch":"refs/heads/master"},"a0d1b66f6a20ca5ebefb6312186205ff57146b94":{"kind":"REWORK","_number":12,"created":"2024-02-05 12:58:33.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/12"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-05 12:59:30.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix : libressl3.4 packages removed\n\nfix: publish docker image jobs added\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a0d1b66f6a20ca5ebefb6312186205ff57146b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a0d1b66f6a20ca5ebefb6312186205ff57146b94"}]},"branch":"refs/heads/master"},"106fcd54fca20532010babe6bf60d2ceb10408d2":{"kind":"REWORK","_number":13,"created":"2024-02-06 08:44:02.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/13"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-06 08:44:54.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix : libressl3.4 packages removed\n\nfix: publish docker image jobs added\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/106fcd54fca20532010babe6bf60d2ceb10408d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/106fcd54fca20532010babe6bf60d2ceb10408d2"}]},"branch":"refs/heads/master"},"9a9fd70fb6544ca40034c8bef6e09fa07ba48b76":{"kind":"REWORK","_number":14,"created":"2024-02-06 08:49:15.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/14"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-06 08:50:13.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix : libressl3.4 packages removed\n\nfix: publish docker image jobs added\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/9a9fd70fb6544ca40034c8bef6e09fa07ba48b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/9a9fd70fb6544ca40034c8bef6e09fa07ba48b76"}]},"branch":"refs/heads/master"},"86f83bbfb30dd7ef37e8c7266ac5ded5d12a3f0c":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2024-02-06 12:53:50.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/15"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-06 12:53:50.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix: libressl3.4 packages removed\n\nfix: publish docker image jobs added\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/86f83bbfb30dd7ef37e8c7266ac5ded5d12a3f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/86f83bbfb30dd7ef37e8c7266ac5ded5d12a3f0c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d9d071b87c37f7406230d68c4bc4a757fc0832ea":{"kind":"REWORK","_number":16,"created":"2024-02-07 05:59:16.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/16"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-07 06:00:06.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix: libressl3.4 packages removed\n\nfix: publish docker image jobs added\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d9d071b87c37f7406230d68c4bc4a757fc0832ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d9d071b87c37f7406230d68c4bc4a757fc0832ea"}]},"branch":"refs/heads/master"},"fe601fa51b0e5887f9557dab0765b4e415b418eb":{"kind":"REWORK","_number":17,"created":"2024-02-08 16:11:29.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/97/907497/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/97/907497/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/97/907497/17"}}},"commit":{"parents":[{"commit":"cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7","subject":"Fix zuul config error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/cdc03f9ae5aeda0a8409e8d555efa488cfb25fa7"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-01 20:38:15.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-08 16:12:23.000000000","tz":0},"subject":"fix: Upgrade librdkafka","message":"fix: Upgrade librdkafka\n\nfix: libressl3.4 packages removed\n\nfix: publish docker image jobs added\n\nChange-Id: Ia8fccfb28b1c95d8ffdb8ca40fba9cc586e5ce7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/fe601fa51b0e5887f9557dab0765b4e415b418eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/fe601fa51b0e5887f9557dab0765b4e415b418eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
