)]}'
{"id":"openstack%2Fmonasca-common~909861","triplet_id":"openstack%2Fmonasca-common~master~I593d0f10b24bc17d3a65927bdb4ae81b2c0acb8a","project":"openstack/monasca-common","branch":"master","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2024-02-22 13:06:31.000000000","reason":"removed on reply"},"35988":{"account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"last_update":"2024-02-22 13:36:20.000000000","reason":"\u003cGERRIT_ACCOUNT_35988\u003e replied on the change","reason_account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}}},"hashtags":[],"change_id":"I593d0f10b24bc17d3a65927bdb4ae81b2c0acb8a","subject":"Adding docker publish as a periodic job","status":"MERGED","created":"2024-02-22 12:54:18.000000000","updated":"2024-02-22 14:17:32.000000000","submitted":"2024-02-22 14:17:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909861","meta_rev_id":"3555d11e03cd68e22d5025b834f742b344614972","_number":909861,"virtual_id_number":909861,"owner":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 14:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":2,"date":"2024-02-22 13:06:31.000000000","_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":1,"date":"2024-02-22 13:36:20.000000000","_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 12:54:38.000000000","updated_by":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"},{"updated":"2024-02-22 13:32:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1596af02e4393852b61f3c9b242f3bd728b0e9fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-22 12:54:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a17fe916eba192cda1f487da4eac03279ffa1ab0","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-02-22 13:06:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff389d0f5ec4bbdcb1bb2babf141dc790a77996d","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-02-22 13:07:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"750674db15797dfd4b0208edacbf2291374e0c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 13:32:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45ffc88b28654fb2b8ae5f032cae4e14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0163b9e1d884a60bea0873f1e6a118a : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53acb63ad5994554af3d68b18629ea9f : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adafd592adf74b06b30a52ec789cd8d4 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/158566c321ff4538a71454f9709dfcdc : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8fffa493cbe4819b7f89244a67afa0d : SUCCESS in 4m 09s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/2740f4f3a94a4112a8938c986b55f4bd : FAILURE in 36m 38s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/a419d8284e7e4547a388897078ed69a8 : FAILURE in 14m 16s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/fb395593fc424d409cb3ad4ee3b75ea7 : SUCCESS in 3m 45s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/2e92b9e1eef64321b197f613d85329fc : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"315315a55a5d71e7a8c7b11e982a6e0177b0b143","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-02-22 13:36:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d570f958c774239f0b019d91aca53102ab4713c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 13:37:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"566bc75a082b44d6e4fb587a8962ed24607b776c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 14:17:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca013646ae74744b0c5a9ba537f7c9e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b93ad2f04545d2b1a4e029931d961b : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36dd9263ce5e413597e321e5f1c9e211 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b37019eb5a6b4203a33c67f5c7e889f3 : SUCCESS in 4m 48s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ff357b703b51401689f65fc4849cb407 : FAILURE in 39m 03s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/c6e142cd27bc472c81494f83ad155268 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3555d11e03cd68e22d5025b834f742b344614972","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 14:17:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f71e5688df41e40c0c57989ee1c31e5bf77d0a2","revisions":{"2f71e5688df41e40c0c57989ee1c31e5bf77d0a2":{"kind":"REWORK","_number":1,"created":"2024-02-22 12:54:18.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/61/909861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/61/909861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/61/909861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/61/909861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/61/909861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/61/909861/1"}}},"commit":{"parents":[{"commit":"fe601fa51b0e5887f9557dab0765b4e415b418eb","subject":"fix: Upgrade librdkafka","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/fe601fa51b0e5887f9557dab0765b4e415b418eb"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-22 12:54:51.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-02-22 12:54:51.000000000","tz":0},"subject":"Adding docker publish as a periodic job","message":"Adding docker publish as a periodic job\n\nChange-Id: I593d0f10b24bc17d3a65927bdb4ae81b2c0acb8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2f71e5688df41e40c0c57989ee1c31e5bf77d0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2f71e5688df41e40c0c57989ee1c31e5bf77d0a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
