)]}'
{"id":"openstack%2Fmonasca-common~920611","triplet_id":"openstack%2Fmonasca-common~master~I5333af865f6c6ef65c575320d47f38408902aa6d","project":"openstack/monasca-common","branch":"master","topic":"feat_kafka_2_4_0","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2024-05-28 09:51:58.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5333af865f6c6ef65c575320d47f38408902aa6d","subject":"upgrade to kafka 2.4.0","status":"MERGED","created":"2024-05-28 08:12:23.000000000","updated":"2024-05-28 11:06:01.000000000","submitted":"2024-05-28 11:06:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920611-feat_kafka_2_4_0","meta_rev_id":"d989f3373ee9783dff94ac5cfce730a1b5e4245b","_number":920611,"virtual_id_number":920611,"owner":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 11:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2024-05-28 09:31:28.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2024-05-28 09:51:58.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-28 09:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7b3e2e6bd47bc3e411b3a393eda112fc7e11848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-28 08:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebea79b4f3cbc8b81bcfc8de7ebacff80d9f82b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:01:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baa8ce6f332e4a17bec0ef2989c7a65d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc41ea49d8b440318193bf691614cfe9 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e002c286db4f4cdaab94c3b958c152e2 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a98344266f5444b88095312084813ba : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fc805ebc4b849f8bf4d6025494e028a : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0c2afa04caf4a21b99942af52753579 : SUCCESS in 3m 31s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/bbe302808bfc48dabc919e1e8c1263b3 : FAILURE in 45m 53s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/b88b66bf0fab402aa4efab487520e9b8 : FAILURE in 15m 20s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/95d88867d94745a6afa647c87ec84bbd : SUCCESS in 3m 31s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/ca4e6cf73d9f40018b601ead7b0527dc : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":1},{"id":"638b8cdf7b3f7a7da1c3646f03af5b82db7ada75","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-28 09:31:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"106ab6f21abb53caed976eafdf45379f7428d31c","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-28 09:43:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0cb9eddd5d6faffbff48e0b2f9f97ec50fb8d7ed","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-28 09:43:37.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_13114\u003e\n","accounts_in_message":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}],"_revision_number":1},{"id":"9e5dcd947826a9661a3f36de3655cbfbfca61df5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:43:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28a75163d9614529bd2f7759f29b2b1175c1e9dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:43:55.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"09fac04bb6ed901d7585f2b46ddf962414e98de9","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-05-28 09:51:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1037657b36f02f98df3a1c83a927695f80c492fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 10:31:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b33edfabe21402caf9cef76804bcbb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63629d7fe8824e5e9f9eedea882666ce : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad31777660824037ae373288d6b4e015 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810e6f5045ef45d49e5e4b1a556bb91c : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35795dd4b3bd4465abf44270ae22331d : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e94e952677743c6836a688bee06056d : SUCCESS in 3m 21s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a73256a2564b4eccb59485b6ac748e4c : FAILURE in 37m 42s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/fa21cc8525334cc79832fb56689e9dd9 : FAILURE in 25m 14s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/9ef9f54ea50f4829b5d9f8b35218009a : SUCCESS in 4m 04s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/2d6c3a5a0ff54e54a50626cedf0ed872 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a425eef3c520e71c7170cbd5fb1b419e87673ff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 10:32:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c51d40fc6bf966115b3137514a9a3727edb3d968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 11:06:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6618dd3e2dfe48b39e1e8d54fd2fa86c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad86d2c2ed84f3392eec44a920514f8 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4be932cd784a44e0b1c3d52586019abe : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a54722e2b4f4cefbdcf187f0d21d6cf : SUCCESS in 3m 15s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/8b623268150a4c33acad054808be3ae8 : FAILURE in 23m 27s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/2efc10de3b0f40e49fa8b988531311f0 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d989f3373ee9783dff94ac5cfce730a1b5e4245b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 11:06:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b18868e8350306a85f89263ece3b3466b351fce7","revisions":{"b18868e8350306a85f89263ece3b3466b351fce7":{"kind":"REWORK","_number":1,"created":"2024-05-28 08:12:23.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/11/920611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/11/920611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/11/920611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/11/920611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/11/920611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/11/920611/1"}}},"commit":{"parents":[{"commit":"2f71e5688df41e40c0c57989ee1c31e5bf77d0a2","subject":"Adding docker publish as a periodic job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2f71e5688df41e40c0c57989ee1c31e5bf77d0a2"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-05-28 08:11:55.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-05-28 08:12:19.000000000","tz":420},"subject":"upgrade to kafka 2.4.0","message":"upgrade to kafka 2.4.0\n\nChange-Id: I5333af865f6c6ef65c575320d47f38408902aa6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
