)]}'
{"id":"openstack%2Fmonasca-common~924031","triplet_id":"openstack%2Fmonasca-common~master~Ic1b5b59341b1841628e98ac1dff022cbeca0706a","project":"openstack/monasca-common","branch":"master","topic":"feat/py312","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-01-08 03:53:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic1b5b59341b1841628e98ac1dff022cbeca0706a","subject":"improve tox","status":"ABANDONED","created":"2024-07-12 05:27:30.000000000","updated":"2025-01-08 03:53:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"65c7ec4198a414a7beaea6a21af079d75060e724","_number":924031,"virtual_id_number":924031,"owner":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-12 11:44:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-12 06:29:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eddf7ba683d0266943923b9c0292b950c8fb043e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-12 05:27:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbac271a3ada524565e081e573fd21b26b892dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-12 05:33:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a59f679c9cd3bcd5cc43886ae05d1558540e8bda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 06:29:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f1204c76e644fc0ac1fa2fb76aa888e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79d313474328450bbcca329ed704536b : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4279721773644ceaa43b51fb6ca368c : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ef0ca7febdd45b984a4d52983a203d8 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4e1608a99004fbf826c7f108e73ac59 : RETRY_LIMIT in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0113c2349c1345f4a1dd111556928937 : SUCCESS in 6m 20s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f8ac3ddd11a14d8c85e39872df261065 : SUCCESS in 54m 12s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/a1f112e75f304a8d895f50bf631cc392 : FAILURE in 36m 42s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/f9c6c6e7d2a142ecaebab21ebcd7e13f : SUCCESS in 3m 09s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/699475f1ef154f12b1b47cb0d162c45e : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2a6651b51a100c23627dcfc6e66aba2df8d419e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-12 10:35:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"28400e4cb70ae1b4c49d004cbc8fdf78955ee9e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-12 10:52:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"455588c8c4be242329725bce0730de7e56a67907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 11:44:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48ed354a3d814c3cb56e207366633ca1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4eb72ed0412f4510a3fe0068ec8b0789 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08396ed703ec43679bde8e32aaf786b1 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f43d6bff3f9549cca40738c793457a44 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e15632d30fc547c9810077fcb2e0070f : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ec0b2100f454dca8dfecd2792edcb36 : SUCCESS in 7m 53s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e9a6285eda3c440da97dd93ed08cbcf7 : SUCCESS in 44m 16s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/3f62737ee1f94ecb92aaa99290024561 : FAILURE in 31m 57s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/8278c1b3f3d64e69891ab5af216208ee : FAILURE in 5m 29s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/07044f00dcfe4c77b2f36e488f4102bc : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":4},{"id":"65c7ec4198a414a7beaea6a21af079d75060e724","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-08 03:53:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3f3d51f6f2ba000f639082b4050f8010888263d0","revisions":{"e6fb35b40603afb2431f4067ddcfa7eb648c2788":{"kind":"REWORK","_number":1,"created":"2024-07-12 05:27:30.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/31/924031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/31/924031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/1"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-12 05:23:12.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-12 05:23:12.000000000","tz":420},"subject":"improve tox","message":"improve tox\n\n- remote whitelist\n- add py312 env\n\nChange-Id: Ic1b5b59341b1841628e98ac1dff022cbeca0706a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e6fb35b40603afb2431f4067ddcfa7eb648c2788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e6fb35b40603afb2431f4067ddcfa7eb648c2788"}]},"branch":"refs/heads/master"},"316b4b3fe24e5b4fbfc4802c8121bd64c7d59a99":{"kind":"REWORK","_number":2,"created":"2024-07-12 05:33:55.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/31/924031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/31/924031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/2"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-12 05:23:12.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-12 05:33:31.000000000","tz":420},"subject":"improve tox","message":"improve tox\n\n- remote whitelist\n- add py312 env\n\nChange-Id: Ic1b5b59341b1841628e98ac1dff022cbeca0706a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/316b4b3fe24e5b4fbfc4802c8121bd64c7d59a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/316b4b3fe24e5b4fbfc4802c8121bd64c7d59a99"}]},"branch":"refs/heads/master"},"14c066a4c53b23770fe63b0a3a807306f4af4638":{"kind":"REWORK","_number":3,"created":"2024-07-12 10:35:31.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/31/924031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/31/924031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/3"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-12 05:23:12.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-12 10:35:18.000000000","tz":420},"subject":"improve tox","message":"improve tox\n\n- remote whitelist\n- add py312 env\n\nChange-Id: Ic1b5b59341b1841628e98ac1dff022cbeca0706a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/14c066a4c53b23770fe63b0a3a807306f4af4638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/14c066a4c53b23770fe63b0a3a807306f4af4638"}]},"branch":"refs/heads/master"},"3f3d51f6f2ba000f639082b4050f8010888263d0":{"kind":"REWORK","_number":4,"created":"2024-07-12 10:52:26.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/31/924031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/31/924031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/31/924031/4"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-12 05:23:12.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-12 10:52:14.000000000","tz":420},"subject":"improve tox","message":"improve tox\n\n- remote whitelist\n- add py312 env\n\nChange-Id: Ic1b5b59341b1841628e98ac1dff022cbeca0706a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3f3d51f6f2ba000f639082b4050f8010888263d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3f3d51f6f2ba000f639082b4050f8010888263d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
