)]}'
{"id":"openstack%2Fmonasca-common~930318","triplet_id":"openstack%2Fmonasca-common~master~I0d147f63fb91d7d455c8e6948fb3854dc63ab29e","project":"openstack/monasca-common","branch":"master","topic":"zuulv3","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-22 05:28:29.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-22 05:28:29.000000000","reason":"Change was submitted"},"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-07-21 17:03:04.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}}},"hashtags":[],"change_id":"I0d147f63fb91d7d455c8e6948fb3854dc63ab29e","subject":"zuul: Move legacy jobs to project","status":"MERGED","created":"2024-09-24 13:21:03.000000000","updated":"2025-07-22 05:28:29.000000000","submitted":"2025-07-22 05:28:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930318-zuulv3","meta_rev_id":"dcbc7305e0c3e92bc405554e2e0e94ae14bfef5d","_number":930318,"virtual_id_number":930318,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-22 05:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2025-07-21 17:03:04.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-21 11:36:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2025-07-22 04:31:50.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-24 14:28:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-30 11:29:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-07-21 11:36:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-07-21 11:36:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"state":"REVIEWER"}],"messages":[{"id":"2758a21bdcc829738137f67b7d04aa7d0412c4d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-24 13:21:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba3888b31cf2e02196606ae1d1d9804337d67186","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-24 13:21:53.000000000","message":"Topic set to zuulv3","accounts_in_message":[],"_revision_number":1},{"id":"f978093eaf24b574f6987b82f3f2975d7ce54ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 14:28:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d445999fb45f489ca39230131118b3f1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b5ac3d7bc0843b289284024d3a2ca7c : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e8f533ffba42129799e9e049094ce9 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/635f6653a3234513b5e29ac9a5272836 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0392eecb64f4353a4b9ecb0b08aee8f : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ac6ce356db74d5ca0607b54c1eaa224 : SUCCESS in 3m 47s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/9db2fbf2ede94ae7bb926187c5dda0fe : SUCCESS in 1h 00m 44s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/7047fb0100a443c88406e035f962de4a : FAILURE in 48m 24s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/35ca1ec1a0b448349ec3fa77590b983d : SUCCESS in 3m 54s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/e210acec6b704e77aba05fc17e180793 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d447d7fc5e471e48184ac0fb113ca05d246e336c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-30 11:28:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7bbbaa5026bdf92bfc021d45525f798a822c155d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-30 11:29:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71d1439f8c3bf5e859079981b863e99cd8e6e309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 12:05:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d90f459c0ea4f209d1d66d8674b075d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d6a49669a944b6ba9d7f64ba6e7b40e : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d862de488f749159a03eae1304e44d5 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/292f2032054e486aa1798fdfbe258731 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e46c8b7522a437a888d43af2e686a8d : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4214a4e5a7a42f3b7d079469a2a5758 : SUCCESS in 5m 31s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/5d295c4f0cfa42f9bff13ed1b611f587 : FAILURE in 22m 49s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/b838de63936f4566a27570d775fb4a97 : FAILURE in 31m 17s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/f7d1c51227f64e18aeecede6f7be3aac : SUCCESS in 3m 56s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/fe5d4d971b554a5381d8bd80093d5d93 : SUCCESS in 1m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72789a5254c694b7ae2ea6985feb9f9c107a01e3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 11:36:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4dc13f12872c2ee8d234b5f823f8fdfef2b90bda","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-21 17:03:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"88046f815eeb30b421327a79d9ef2a0f9752bf54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 17:05:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09877d243099aad4f32e3fa7dbced60d94ac4b91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 17:51:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6cb04b8462d4e6bb24634fac7bb1494\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4ce109addc4940bcbdf065d5f75e49 : SUCCESS in 2m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2413de33f33640edb7f1d3b889f8bbeb : FAILURE in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8e870c7491c4c0592efcfa5bdf393a6 : SUCCESS in 3m 05s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f4f19eb409474bd2ab67b2bccc273872 : FAILURE in 33m 54s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/200030f40957437e8b9d9de0fcc9f1f2 : SUCCESS in 23s","accounts_in_message":[],"_revision_number":2},{"id":"fa59839d6b353dc94e6a88310006181b4d8459cb","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-22 04:21:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea66d1d72d850991c253aea1f34b2ccba20e480a","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-22 04:31:45.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"48a3d1ad5dc7dc1a6ce6f6099770094523a291bc","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-22 04:31:50.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d8399e3254e7772790e975d94a32958592f72b79","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-22 04:32:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04f94446b66a93e6894edc6433bb834904477afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 04:54:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea87ef4fd1e043e085540d11b3444a24\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35ae93efa4734cca86ca122206d60a40 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c788c51bf9134f30928355c84cf55106 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02d3a5676eed4cab80541da8ec14a1d4 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ba249c0f2c64035a88a4b3794d8f996 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b13605c1dbd64054b08d5349d03a4f54 : SUCCESS in 4m 01s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/66283bf52c27470f84d72e557ffc1843 : FAILURE in 30m 34s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/042bb331d7114aa3adcdad7b234bf40b : FAILURE in 17m 26s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/00bae277b7f64c679e2cbc4de489f144 : SUCCESS in 27s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/095a0701a113449cbc62b1da00959888 : SUCCESS in 1m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8339717ce06285b72d41b5e7699c17e973b30447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 04:54:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4bdfbbad1dd1e54b34cae86c12916159e3b4136c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 05:28:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93dddd3a513343a28d3d09020e6735fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eff10c1a23f4210b1c6575f8d00645e : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c637389dbb2f4e7091c53b126634cf40 : SUCCESS in 7m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25bdd4540ddd44459794dfd6d77abbe2 : SUCCESS in 3m 02s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/79c901fe6c274d1886d36c71646abd40 : FAILURE in 29m 25s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/03bbf68115d54add8084b6fe53769cfd : SUCCESS in 26s","accounts_in_message":[],"_revision_number":2},{"id":"dcbc7305e0c3e92bc405554e2e0e94ae14bfef5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 05:28:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e875708316cfe24dbf300f0df135fbc98da607aa","revisions":{"ddde3f2088abf674b048124864e7541c579d315d":{"kind":"REWORK","_number":1,"created":"2024-09-24 13:21:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/930318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/18/930318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/930318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/930318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/930318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/18/930318/1"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-24 13:20:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-24 13:20:55.000000000","tz":60},"subject":"zuul: Move legacy jobs to project","message":"zuul: Move legacy jobs to project\n\nThese jobs need to be rewritten for zuulv3, but for now get them out of\nthe openstack-zuul-jobs repo.\n\nChange-Id: I0d147f63fb91d7d455c8e6948fb3854dc63ab29e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ddde3f2088abf674b048124864e7541c579d315d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ddde3f2088abf674b048124864e7541c579d315d"}]},"branch":"refs/heads/master"},"e875708316cfe24dbf300f0df135fbc98da607aa":{"kind":"REWORK","_number":2,"created":"2025-04-30 11:28:45.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/18/930318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/18/930318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/930318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/930318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/930318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/18/930318/2"}}},"commit":{"parents":[{"commit":"87c25171df04c1885a971119a5dcb608d7327450","subject":"upgrade libkafka 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-24 13:20:55.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-04-30 11:28:20.000000000","tz":120},"subject":"zuul: Move legacy jobs to project","message":"zuul: Move legacy jobs to project\n\nThese jobs need to be rewritten for zuulv3, but for now get them out of\nthe openstack-zuul-jobs repo.\n\nChange-Id: I0d147f63fb91d7d455c8e6948fb3854dc63ab29e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e875708316cfe24dbf300f0df135fbc98da607aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e875708316cfe24dbf300f0df135fbc98da607aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
