)]}'
{"id":"openstack%2Fmonasca-common~938350","triplet_id":"openstack%2Fmonasca-common~master~Ifaecf9039581cd3851b47344f71ad2c1c390484d","project":"openstack/monasca-common","branch":"master","topic":"upgrade_monasca","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-01-08 05:15:54.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}}},"hashtags":[],"change_id":"Ifaecf9039581cd3851b47344f71ad2c1c390484d","subject":"upgrade monasca","status":"MERGED","created":"2025-01-02 07:37:17.000000000","updated":"2025-01-08 05:57:21.000000000","submitted":"2025-01-08 05:57:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938350-upgrade_monasca","meta_rev_id":"5e971660451cd70ff794c671af98cb6121f006ec","_number":938350,"virtual_id_number":938350,"owner":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-08 05:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2025-01-08 05:15:54.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2025-01-08 05:15:54.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-02 08:24:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42ead6c60dcfad894562c92b57b29caa3a1c76ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-02 07:37:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61c39930cd42b7bb497a7dec7d9be0f701887df5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-02 07:44:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a55a52743b3e881268de9405b27fbb26c23d53d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 08:24:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9989adf4b2e41619ae128db4413e390\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaab071a9487471084e11bb03e7c7bce : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bb59234c4c417986d47a70ec47e155 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d3a8031d69d4f1e9a475cb220567f79 : FAILURE in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2552f007131450bbf3775b5923b33d7 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/869e367e14f8432691f3b32360b9f32b : SUCCESS in 4m 21s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/13253016f4824049bb00298ca4ea2f1e : FAILURE in 33m 38s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/66dcea755f8d43b98289e5c2c80f14f2 : FAILURE in 19m 38s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/97b90fb5c70645a18d60861a991c38ce : SUCCESS in 3m 45s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/f0f327ca352b40168db4094fb0bbda36 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b9fae74e4d30889c0670f01fcdad13e3afa6556d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-08 03:58:00.000000000","message":"Topic maven_https removed","accounts_in_message":[],"_revision_number":2},{"id":"3db35dc80d1289e235ed3e36a3c24008b729e1a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-08 04:00:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"32a0482f40595ba64609a7011fc3ad408c3bbbbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 04:00:44.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/monasca-common on branch master.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/monasca-common/.zuul.yaml@master\", line 12, column 19","accounts_in_message":[],"_revision_number":3},{"id":"8a280c2111c3e7f4f4e9a03d58a0b146b3694aa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-08 04:06:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"95a2864fc03f3219f1f1e8c6097818d14c9f3307","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-08 04:08:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"108b4d270f5ad50049e041bc7511b8d8c7ed75a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 04:46:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f6b758a8d8d4656a4c352fc8cbadec4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c11c6e43e1a34e828c5c906986d7a01d : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba4783493d0e4f86b0c1655ca04de288 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3359a2e018247c6a1b2832bc4143394 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6262578aeeb64b5cb05a36a094fabb96 : SUCCESS in 4m 06s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/cbd82f25aa194e7197d6401974f29800 : FAILURE in 32m 42s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/7c7c97215705479fa9b1764571317d07 : FAILURE in 30m 39s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/0552de64466e47a7a48c439766ddb707 : SUCCESS in 3m 10s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/803be13cff8c479385e9d537e0ab895b : SUCCESS in 2m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"befb14ade9a78923fa48895edb2cc49ce2f80874","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-08 04:54:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e9e532383158826dec354147c0257041dd3d6c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 05:11:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9cd650940974d6b974b7fc80085b6a6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b14a4df2a274bc0b3bf81cbcf024082 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f207ca7c794439871fd8dff86acba9 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c148c137dc3d4b8f9e1e7bff69ea820d : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fb76125c2174294b8ab91f8ef2a2c69 : SUCCESS in 3m 00s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/4b5af5acccb742749d8eee4164789715 : FAILURE in 14m 39s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/e812fb9f72f44ab1a2f8af12f0173d18 : FAILURE in 16m 14s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/9be0579e0fe344e9a0ab6835f3b4b906 : SUCCESS in 3m 25s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/9ff18276051a48f297ab65e209d01a03 : SUCCESS in 1m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f98a0f03c90a7c71239a1fa0e679f57e57c4b401","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-01-08 05:15:54.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4d8cf16b6a77be2e19a3001a59b2d85fb3faa013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 05:16:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ceff126af06d57c0470b17da788480e6ff0715ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 05:57:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/287efce0c06f4577adea5a34bb885363\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d631d5f89b484432bdfedec3b1bcf196 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5d2e19484e1462faa3f776c1c7d1826 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45e7b00a7f5a4ccfa317e45c2fd6bc2e : SUCCESS in 3m 01s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/46e6dd86baec4d0285a817089887a7d6 : FAILURE in 34m 42s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/b49b7151c3ab4656a29657dab61561b9 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":6},{"id":"5e971660451cd70ff794c671af98cb6121f006ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 05:57:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a723f69d59fe0ac2dbe3d0ba31b104e309f2e61b","revisions":{"05bdac4624d474dd293b084f3ef4516f10ae533e":{"kind":"REWORK","_number":1,"created":"2025-01-02 07:37:17.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/50/938350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/50/938350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/1"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-01-02 07:37:07.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-01-02 07:37:07.000000000","tz":420},"subject":"upgrade maven and use https repo","message":"upgrade maven and use https repo\n\nChange-Id: Ifaecf9039581cd3851b47344f71ad2c1c390484d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/05bdac4624d474dd293b084f3ef4516f10ae533e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/05bdac4624d474dd293b084f3ef4516f10ae533e"}]},"branch":"refs/heads/master"},"458dcbddfa10b9fb966364b9d34627fcad993313":{"kind":"REWORK","_number":2,"created":"2025-01-02 07:44:17.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/50/938350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/50/938350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/2"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-01-02 07:37:07.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-01-02 07:44:00.000000000","tz":420},"subject":"upgrade maven and use https repo","message":"upgrade maven and use https repo\n\nChange-Id: Ifaecf9039581cd3851b47344f71ad2c1c390484d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/458dcbddfa10b9fb966364b9d34627fcad993313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/458dcbddfa10b9fb966364b9d34627fcad993313"}]},"branch":"refs/heads/master"},"94d71acb3bfc4435852973d149d159c1c16e26c9":{"kind":"REWORK","_number":3,"created":"2025-01-08 04:00:24.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/50/938350/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/50/938350/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/3"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-01-02 07:37:07.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh@fpt.com","date":"2025-01-08 03:59:32.000000000","tz":420},"subject":"upgrade monasca","message":"upgrade monasca\n\n- upgrade maven and use https repo\n- temporary disable some jobs\n\nChange-Id: Ifaecf9039581cd3851b47344f71ad2c1c390484d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/94d71acb3bfc4435852973d149d159c1c16e26c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/94d71acb3bfc4435852973d149d159c1c16e26c9"}]},"branch":"refs/heads/master"},"d43ec785fdd2df782abf8509187f3d01df5c25db":{"kind":"REWORK","_number":4,"created":"2025-01-08 04:06:55.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/50/938350/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/50/938350/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/4"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-01-02 07:37:07.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh@fpt.com","date":"2025-01-08 04:06:43.000000000","tz":420},"subject":"upgrade monasca","message":"upgrade monasca\n\n- upgrade maven and use https repo\n- temporary disable some jobs\n\nChange-Id: Ifaecf9039581cd3851b47344f71ad2c1c390484d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d43ec785fdd2df782abf8509187f3d01df5c25db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d43ec785fdd2df782abf8509187f3d01df5c25db"}]},"branch":"refs/heads/master"},"2c7a6affe5deba864f64ad1882e5461f4a3d9b34":{"kind":"REWORK","_number":5,"created":"2025-01-08 04:08:31.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/50/938350/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/50/938350/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/5"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-01-02 07:37:07.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh@fpt.com","date":"2025-01-08 04:08:17.000000000","tz":420},"subject":"upgrade monasca","message":"upgrade monasca\n\n- upgrade maven and use https repo\n- temporary disable some jobs\n\nChange-Id: Ifaecf9039581cd3851b47344f71ad2c1c390484d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2c7a6affe5deba864f64ad1882e5461f4a3d9b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2c7a6affe5deba864f64ad1882e5461f4a3d9b34"}]},"branch":"refs/heads/master"},"a723f69d59fe0ac2dbe3d0ba31b104e309f2e61b":{"kind":"REWORK","_number":6,"created":"2025-01-08 04:54:07.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/50/938350/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/50/938350/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/50/938350/6"}}},"commit":{"parents":[{"commit":"b18868e8350306a85f89263ece3b3466b351fce7","subject":"upgrade to kafka 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b18868e8350306a85f89263ece3b3466b351fce7"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-01-02 07:37:07.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh@fpt.com","date":"2025-01-08 04:49:55.000000000","tz":420},"subject":"upgrade monasca","message":"upgrade monasca\n\n- upgrade maven and use https repo\n- temporary disable some jobs\n- upgrade librdkafka\n\nChange-Id: Ifaecf9039581cd3851b47344f71ad2c1c390484d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a723f69d59fe0ac2dbe3d0ba31b104e309f2e61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a723f69d59fe0ac2dbe3d0ba31b104e309f2e61b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
