)]}'
{"id":"openstack%2Fmonasca-common~943109","triplet_id":"openstack%2Fmonasca-common~master~I2ddbe239a3ca262ba8c849bd07d65b0fbc8f56a1","project":"openstack/monasca-common","branch":"master","hashtags":[],"change_id":"I2ddbe239a3ca262ba8c849bd07d65b0fbc8f56a1","subject":"Fix: Dummy Change","status":"ABANDONED","created":"2025-03-01 22:06:27.000000000","updated":"2025-03-01 23:15:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4a50a7511ae662a8eb78d99422627f2a459fe0ca","_number":943109,"virtual_id_number":943109,"owner":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-01 22:14:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-01 22:14:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b98a9f85283d26756698996f64704f511381287","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2025-03-01 22:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf92a1ba787ea394bb593ca19e751f45df4f9c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 22:14:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1d1697b72c4419afe96052fe870164\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f9bcaa71d4040929782e0f189228924 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/733be798cb7b42edb489f3f58cd68850 : SUCCESS in 3m 15s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/63641759ee944a429bf1b079c9e1819e : SUCCESS in 3m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a50a7511ae662a8eb78d99422627f2a459fe0ca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2025-03-01 23:15:37.000000000","message":"Abandoned\n\njust a test commit","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6e8ebb832a0cca9272cd2bea815bfff48304ad8","revisions":{"c6e8ebb832a0cca9272cd2bea815bfff48304ad8":{"kind":"REWORK","_number":1,"created":"2025-03-01 22:06:27.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/09/943109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/09/943109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/943109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/943109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/943109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/09/943109/1"}}},"commit":{"parents":[{"commit":"a723f69d59fe0ac2dbe3d0ba31b104e309f2e61b","subject":"upgrade monasca","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a723f69d59fe0ac2dbe3d0ba31b104e309f2e61b"}]}],"author":{"name":"Hasan ACAR","email":"hasan.acar@tubitak.gov.tr","date":"2025-03-01 22:06:14.000000000","tz":0},"committer":{"name":"Hasan ACAR","email":"hasan.acar@tubitak.gov.tr","date":"2025-03-01 22:06:14.000000000","tz":0},"subject":"Fix: Dummy Change","message":"Fix: Dummy Change\n\nChange-Id: I2ddbe239a3ca262ba8c849bd07d65b0fbc8f56a1\nSigned-off-by: Hasan ACAR \u003chasan.acar@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/c6e8ebb832a0cca9272cd2bea815bfff48304ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/c6e8ebb832a0cca9272cd2bea815bfff48304ad8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
