)]}'
{"id":"openstack%2Fmonasca-common~946029","triplet_id":"openstack%2Fmonasca-common~master~I8afd8add9183271d0329bc0b0c88954caf93d822","project":"openstack/monasca-common","branch":"master","topic":"upgrade_monasca","hashtags":[],"change_id":"I8afd8add9183271d0329bc0b0c88954caf93d822","subject":"upgrade libkafka 2.8.0","status":"MERGED","created":"2025-04-01 11:48:14.000000000","updated":"2025-04-01 13:25:04.000000000","submitted":"2025-04-01 13:25:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946029-upgrade_monasca","meta_rev_id":"72659ecc24bb4784f695ae624d6faf137990f1e7","_number":946029,"virtual_id_number":946029,"owner":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-01 13:25:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2025-04-01 12:55:32.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2025-04-01 12:55:32.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-01 12:23:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfb7bdef0b210f3db01eb13db2f6f161bd55d55f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-04-01 11:48:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ba1436c7993ce804adf41d59b2d674d1efb20fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 12:23:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd3888699dd24537bfec1b1b1c1867cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8e44aaacf8d481290ebda48288855c8 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c57b5e9e2ea44a79c1981b44b51da2b : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/202f6cdbdb8843198abbcddf00c9f795 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/beed0792d82c451d8dde1a035effc4c7 : SUCCESS in 3m 17s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ba4d6138e0c74417bf171a0450305698 : FAILURE in 18m 48s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/d3464a90935a4dc883d47897e1a3bb2b : FAILURE in 25m 59s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/3b138ffde6a640dab6759b47a2705945 : SUCCESS in 3m 12s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/0976b5c32ead42b7b6b57897ac3ea2a1 : SUCCESS in 1m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d02602747c3694c4b32e692fc7367b9c37a53a81","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-04-01 12:55:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4502df05e873406afcec4a7b7c0818ce3b34c31c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 12:56:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4c203e44da38abfc6a16864fe0614839c7e6fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 13:25:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd786a1a07ab40d2a793b5323931388a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bff54f9d4b4ea28fdb303dc2f27b63 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3084432f52334b7cb4fa97640436a2c5 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6be5a47066b249b08f329da51c653eca : SUCCESS in 3m 12s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/19586e69d805465ab88a37cf4cc0a809 : FAILURE in 24m 48s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/a060567998444396a35904151a7e57ba : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"72659ecc24bb4784f695ae624d6faf137990f1e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 13:25:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87c25171df04c1885a971119a5dcb608d7327450","revisions":{"87c25171df04c1885a971119a5dcb608d7327450":{"kind":"REWORK","_number":1,"created":"2025-04-01 11:48:14.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/29/946029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/29/946029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/29/946029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/29/946029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/29/946029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/29/946029/1"}}},"commit":{"parents":[{"commit":"a723f69d59fe0ac2dbe3d0ba31b104e309f2e61b","subject":"upgrade monasca","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a723f69d59fe0ac2dbe3d0ba31b104e309f2e61b"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-04-01 11:48:04.000000000","tz":0},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2025-04-01 11:48:04.000000000","tz":0},"subject":"upgrade libkafka 2.8.0","message":"upgrade libkafka 2.8.0\n\nChange-Id: I8afd8add9183271d0329bc0b0c88954caf93d822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
