)]}'
{"id":"openstack%2Fmonasca-common~955509","triplet_id":"openstack%2Fmonasca-common~master~Ia3d6368a6512ac17996bf478f355ddbe795d5808","project":"openstack/monasca-common","branch":"master","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-07-22 03:54:58.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-22 04:10:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3d6368a6512ac17996bf478f355ddbe795d5808","subject":"Fix mistake in RequestContext","status":"MERGED","created":"2025-07-21 18:57:47.000000000","updated":"2025-07-22 04:10:36.000000000","submitted":"2025-07-22 04:10:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955509","meta_rev_id":"253c78fd2e8ff465ddfcd1c5a837fb19843c2eec","_number":955509,"virtual_id_number":955509,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-22 04:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2025-07-22 03:54:58.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2025-07-22 03:54:58.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-21 19:39:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-21 21:01:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"state":"REVIEWER"}],"messages":[{"id":"2b323bdf1bf190b548dc7ffa0eebb883ceb10d9f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 18:57:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0fd14e5d49768c654e39ea9cb158725febcd611","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 18:58:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4ee8cf4a89e77856f2148859954c8b54c45e2ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 19:39:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd273c349a064bdcbc9a78ab2d8137a3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2b689caad714aa1b67033b26a164437 : FAILURE in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69585e85b4804bc4a5468b222a6f2242 : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9bb385f48784cf3b8712121152a8c4d : FAILURE in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e110fd89b4b405e8e860c3e044deb9f : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04673f110f8242e8bbdad2b61d4b1af4 : SUCCESS in 5m 16s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/5df2f03161474bc1a82c36bbfbce7230 : FAILURE in 28m 22s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/5bea04d0dfbb4b5291db7cb0b9ccabb3 : FAILURE in 12m 40s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/9b04af67f8e24f009de335a076c41c2f : SUCCESS in 23s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/5f8422e366524f72bd0cad653271bc8b : SUCCESS in 1m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2159cecdbaf8450084e006801d0dfa62f0c7083e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 20:12:00.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d8506d6652c681d8f30fa642e16f2dbd2c601f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 20:45:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1cfe1ccadcb4021a83261a20e1c187b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f14dbef046ea4080a0ebb0931e905cf2 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b58a87694b8442eaa8f871a82d2938aa : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d06842925c6471d9effe639fe67c01f : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/448ac15cb5d04d90a04bdb9abc70586a : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58607cf30c304f8b9dddeb7ea3deb74a : SUCCESS in 5m 03s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/599275c76d394666afcb8cc30f47fab8 : FAILURE in 12m 06s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/70ced2bb49b24b818bc500183a85a495 : FAILURE in 31m 47s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/22d83f2a094c451cae5645d11ab33870 : SUCCESS in 26s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/cef1b849aef7444b9eff0dca69db6418 : SUCCESS in 1m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b18991b5c65e7536c056854b1afc828771b22325","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 21:01:33.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"86b635cafa8183875d391a4b0ef4be03ae0a872d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 21:01:46.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"861b56010739b4590e7c8c5951293ae031d9abc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 21:03:02.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0af7b74415f2f175983792be3d25fa3850e1baa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-21 21:03:33.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"eec8e989cbdfcce156f7d365189a8ec43b287050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 21:42:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0557ba8f57a14b7abc62547e8b4a1fc6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a860e8177b140f8a0450559dc5ed0fd : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a063c67d6fad48a68ffb79446fa382d7 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0d3eeca209d46b8a37e2d337eebd51c : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c6c01aef110485fb1fe4b0a7fe28b94 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53eae7675c214f95985c0f886f2e002e : SUCCESS in 5m 20s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a51db387fe784609a71bbcb41a43d924 : FAILURE in 32m 33s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/2da2161e61f54f2ca768d7f826c4fbde : FAILURE in 29m 58s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/c15093e331234ea78a435346a84c1e74 : SUCCESS in 28s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/41a1dea984f74302a4fc88807b704e72 : SUCCESS in 1m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"db3df6337ee51254a726122e8b83f404fee2ffca","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-22 03:54:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b744df68854155da77f0ef3c6c20fbf8fcd14aec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 03:55:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4e424d19bfe6a7f1ae4666ac2b9f3ec328fc445f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 04:10:36.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d131e620b04e4a6c96d79b1bdc455fef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2314efb62fd49e39a9083cca7282c5c : SUCCESS in 1m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19b20fc107c44f34b027be220a66b293 : SUCCESS in 2m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f00e64ab465d4253b534b193afdef626 : SUCCESS in 1m 58s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e296c7b913654d1e91135bb6d274cb01 : FAILURE in 14m 22s (non-voting)\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/6d8a14c05a394e21baea41a346c36b0c : SUCCESS in 24s","accounts_in_message":[],"_revision_number":6},{"id":"253c78fd2e8ff465ddfcd1c5a837fb19843c2eec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 04:10:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"800465f295aa915a656df7c6ed0975e6c6f76c7c","revisions":{"97e8378c6942147a8fb2b1a0793d4bf361a8ea98":{"kind":"REWORK","_number":1,"created":"2025-07-21 18:57:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/955509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/09/955509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/1"}}},"commit":{"parents":[{"commit":"87c25171df04c1885a971119a5dcb608d7327450","subject":"upgrade libkafka 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 18:57:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 18:57:47.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ia3d6368a6512ac17996bf478f355ddbe795d5808\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/97e8378c6942147a8fb2b1a0793d4bf361a8ea98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/97e8378c6942147a8fb2b1a0793d4bf361a8ea98"}]},"branch":"refs/heads/master"},"496d391a60ce91f1edb455c7ff99a5fa409e02f9":{"kind":"REWORK","_number":2,"created":"2025-07-21 18:58:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/955509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/09/955509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/2"}}},"commit":{"parents":[{"commit":"87c25171df04c1885a971119a5dcb608d7327450","subject":"upgrade libkafka 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 18:57:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 18:58:35.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ia3d6368a6512ac17996bf478f355ddbe795d5808\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/496d391a60ce91f1edb455c7ff99a5fa409e02f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/496d391a60ce91f1edb455c7ff99a5fa409e02f9"}]},"branch":"refs/heads/master"},"b3639c9d652a01cebaf437d8edb4c002b333519d":{"kind":"REWORK","_number":3,"created":"2025-07-21 20:12:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/955509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/09/955509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/3"}}},"commit":{"parents":[{"commit":"87c25171df04c1885a971119a5dcb608d7327450","subject":"upgrade libkafka 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 18:57:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 20:12:00.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ia3d6368a6512ac17996bf478f355ddbe795d5808\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b3639c9d652a01cebaf437d8edb4c002b333519d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b3639c9d652a01cebaf437d8edb4c002b333519d"}]},"branch":"refs/heads/master"},"a824872a2a8de32098a90b5650cedb17ab08380d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-07-21 21:01:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/955509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/09/955509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/4"}}},"commit":{"parents":[{"commit":"87c25171df04c1885a971119a5dcb608d7327450","subject":"upgrade libkafka 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 18:57:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 21:01:33.000000000","tz":0},"subject":"Fix mistakes in tests","message":"Fix mistakes in tests\n\nChange-Id: Ia3d6368a6512ac17996bf478f355ddbe795d5808\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a824872a2a8de32098a90b5650cedb17ab08380d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a824872a2a8de32098a90b5650cedb17ab08380d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b2c3fb740cfdfbef83e81be0c073ced05ec7be14":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-07-21 21:03:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/955509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/09/955509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/5"}}},"commit":{"parents":[{"commit":"87c25171df04c1885a971119a5dcb608d7327450","subject":"upgrade libkafka 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 18:57:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 21:03:02.000000000","tz":0},"subject":"Fix mistakes in test","message":"Fix mistakes in test\n\nChange-Id: Ia3d6368a6512ac17996bf478f355ddbe795d5808\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b2c3fb740cfdfbef83e81be0c073ced05ec7be14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b2c3fb740cfdfbef83e81be0c073ced05ec7be14"}]},"branch":"refs/heads/master","description":"Edit commit message"},"800465f295aa915a656df7c6ed0975e6c6f76c7c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-07-21 21:03:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/955509/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/09/955509/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/09/955509/6"}}},"commit":{"parents":[{"commit":"87c25171df04c1885a971119a5dcb608d7327450","subject":"upgrade libkafka 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/87c25171df04c1885a971119a5dcb608d7327450"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 18:57:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-21 21:03:33.000000000","tz":0},"subject":"Fix mistake in RequestContext","message":"Fix mistake in RequestContext\n\nChange-Id: Ia3d6368a6512ac17996bf478f355ddbe795d5808\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/800465f295aa915a656df7c6ed0975e6c6f76c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/800465f295aa915a656df7c6ed0975e6c6f76c7c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
