)]}'
{"id":"openstack%2Fmonasca-common~955574","triplet_id":"openstack%2Fmonasca-common~master~I795536eb79f19ddcb36ead87387ac49b2f30160c","project":"openstack/monasca-common","branch":"master","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-07-23 05:07:36.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-23 05:43:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I795536eb79f19ddcb36ead87387ac49b2f30160c","subject":"tox: Remove ineffective ignore_basepython_conflict and bump minimum version","status":"MERGED","created":"2025-07-22 09:51:28.000000000","updated":"2025-07-23 05:43:24.000000000","submitted":"2025-07-23 05:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955574","meta_rev_id":"59c63f38263acd3d9d2fbc08361e9e1218fb9b72","_number":955574,"virtual_id_number":955574,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-23 05:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2025-07-23 05:07:36.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2025-07-23 05:07:36.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-22 11:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-22 11:09:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"state":"REVIEWER"}],"messages":[{"id":"02e9877af526251af2693c35edbd01fda55ab903","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 09:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70eb68407828cf67ffe3996fbcd57ec5b0684d9b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 09:54:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cea575aacb0705216faf7a261e4053b90a545433","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 10:17:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f5aabb5504fb3b1fe72f3607050b79c49e23c6cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 11:06:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b2ffcf009b74ed7b74d7ff77055ea0a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/092f600fe01c40e2aff2841a3bcef71f : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c95206218e45aeb5ccd05745830dc6 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3191f2a622624b9cae7d0bb31051e835 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a27776ee74ec4164bc69d35033772519 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59d28364747c4cf38a7bf4d64baa6a74 : SUCCESS in 5m 56s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/210f43452edd40bfbb794f7cb0d3ba8b : FAILURE in 37m 41s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/c4680fd8fb2045bc93620ce9318802b9 : FAILURE in 36m 04s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/e92358f3d5a44dc8aa05a629900e523c : SUCCESS in 25s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/7077497430da4b8ea7f322d137cb5738 : SUCCESS in 1m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c65fae35dc21a09b6620617e272cba502e865903","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 11:09:39.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"b933cd434e9018abebcffb4572b38217dfb41fee","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-23 05:07:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"483744f77a25f5f5fb7243ac50a5ceac10df0eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 05:08:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb74cbaead7261879a21c898ece5c2bcdb21105b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 05:43:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b9c85456364c018b5e59e6f2489729\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00b1abe31a0343e1befda237f4468d4c : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2563a69cafec41c8a162de9fa90b8003 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddc4ca75cec942e1afaa11ea64681a01 : SUCCESS in 2m 55s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a14383cb45354a238ed5077245d7b8f1 : FAILURE in 17m 27s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/e97787c8766b4b4d900adcba1012e895 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":3},{"id":"59c63f38263acd3d9d2fbc08361e9e1218fb9b72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 05:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e440a3354a61c65cd40138772891810d4e54d8cf","revisions":{"610b8f384850f13c4c2ad252ce4c396253a8fd97":{"kind":"REWORK","_number":1,"created":"2025-07-22 09:51:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/74/955574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/74/955574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/1"}}},"commit":{"parents":[{"commit":"bf623ef346852d39c6b6e8ec1449961ecf6cf165","subject":"Merge \"zuul: Move legacy jobs to project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bf623ef346852d39c6b6e8ec1449961ecf6cf165"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 09:51:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 09:51:28.000000000","tz":0},"subject":"tox: Remove ineffective ignore_basepython_conflict and bump minimum version","message":"tox: Remove ineffective ignore_basepython_conflict and bump minimum version\n\nChange-Id: I795536eb79f19ddcb36ead87387ac49b2f30160c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/610b8f384850f13c4c2ad252ce4c396253a8fd97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/610b8f384850f13c4c2ad252ce4c396253a8fd97"}]},"branch":"refs/heads/master"},"555dbb030f1de93704c4f40a21e148ca2ed1afc1":{"kind":"REWORK","_number":2,"created":"2025-07-22 09:54:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/74/955574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/74/955574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/2"}}},"commit":{"parents":[{"commit":"bf623ef346852d39c6b6e8ec1449961ecf6cf165","subject":"Merge \"zuul: Move legacy jobs to project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bf623ef346852d39c6b6e8ec1449961ecf6cf165"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 09:51:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 09:54:56.000000000","tz":0},"subject":"tox: Remove ineffective ignore_basepython_conflict and bump minimum version","message":"tox: Remove ineffective ignore_basepython_conflict and bump minimum version\n\nChange-Id: I795536eb79f19ddcb36ead87387ac49b2f30160c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/555dbb030f1de93704c4f40a21e148ca2ed1afc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/555dbb030f1de93704c4f40a21e148ca2ed1afc1"}]},"branch":"refs/heads/master"},"e440a3354a61c65cd40138772891810d4e54d8cf":{"kind":"REWORK","_number":3,"created":"2025-07-22 10:17:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/74/955574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/74/955574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/74/955574/3"}}},"commit":{"parents":[{"commit":"bf623ef346852d39c6b6e8ec1449961ecf6cf165","subject":"Merge \"zuul: Move legacy jobs to project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bf623ef346852d39c6b6e8ec1449961ecf6cf165"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 09:51:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 10:17:41.000000000","tz":0},"subject":"tox: Remove ineffective ignore_basepython_conflict and bump minimum version","message":"tox: Remove ineffective ignore_basepython_conflict and bump minimum version\n\nChange-Id: I795536eb79f19ddcb36ead87387ac49b2f30160c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e440a3354a61c65cd40138772891810d4e54d8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e440a3354a61c65cd40138772891810d4e54d8cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
