)]}'
{"id":"openstack%2Fmonasca-common~955580","triplet_id":"openstack%2Fmonasca-common~master~Iabdf2595fd9645ce66f3c9377e48074cb71e4cc8","project":"openstack/monasca-common","branch":"master","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-07-23 05:05:59.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-23 05:43:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iabdf2595fd9645ce66f3c9377e48074cb71e4cc8","subject":"Remove deprecated pbr option","status":"MERGED","created":"2025-07-22 10:46:38.000000000","updated":"2025-07-23 05:43:12.000000000","submitted":"2025-07-23 05:43:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955580","meta_rev_id":"9dc1d468f1063056814c7f6d88522781642ea5e9","_number":955580,"virtual_id_number":955580,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-23 05:43:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2025-07-23 05:05:59.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2025-07-23 05:05:59.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-22 10:47:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"state":"REVIEWER"},{"updated":"2025-07-22 11:24:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8b221859b6a5beb6cb2be114b04e4e3ff4b4ba8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 10:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"152368c9d9a2bce982a2135b8cc5c28e499c7ec1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 10:47:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3d95531e34559f46568b5d9f834038807ec5549f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 10:47:33.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"e1a900bb4044be4efd7bfaafccd7ea5170c2e0bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 11:24:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61afb8a43b08476f9333034957a3f521\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d10ec3c058764f66b3d1d94404b306d9 : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39235bbe55d34add98ff6f5d627f9b07 : SUCCESS in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a162085d344ba5b7fd68f224c9a352 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84aa123df5394a8cb20c2af835609baa : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ccf3985f0be42709059570a9c886732 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4916a9d6e30f40c8acce7a2f3fea860b : SUCCESS in 5m 45s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/6b00bef5cf41462b8fb1127aeeec3b8f : FAILURE in 27m 30s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/60729af9d4774487b167777085ab3aca : FAILURE in 32m 19s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/ea0cf68a765040858d66c4e3e0c2e2eb : SUCCESS in 26s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/e2b8a3e3b831463ba54e0876898b8439 : SUCCESS in 2m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b0bd6a9fce1ec98de6f0c10fccaabb02185b4c7","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-23 05:05:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"33269089e561b980afd2dbc0fcc30827558563d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 05:06:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"227d0a135e4c0231c36d59acf016507032586212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 05:43:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1883e05b7a3241ca8f3ae51ba689b582\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/889fce80097640ae829ca95265d7d25e : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e98845674448838c982dce35ff959d : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f66c8753de94be48d366d633dee41d4 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d224c9e681994a67acc14ba6d312a9b6 : SUCCESS in 3m 12s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/a863e45d30714aefa62a476db337f8fc : FAILURE in 24m 26s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/71350e713e6c4673910b18dc4a6f43bd : SUCCESS in 24s","accounts_in_message":[],"_revision_number":2},{"id":"9dc1d468f1063056814c7f6d88522781642ea5e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 05:43:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e31214c54b047f0945f467f590091bae680fa33","revisions":{"e6006cc1ad1f5c85b70c49df4ea6fa4a3f073100":{"kind":"REWORK","_number":1,"created":"2025-07-22 10:46:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/955580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/80/955580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/80/955580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/80/955580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/80/955580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/80/955580/1"}}},"commit":{"parents":[{"commit":"d54f6c655a8738f25cdc70cb0a323f18995f4884","subject":"Remove default override for config options policy_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d54f6c655a8738f25cdc70cb0a323f18995f4884"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 10:46:38.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 10:46:38.000000000","tz":0},"subject":"Remove deprecated pbr option","message":"Remove deprecated pbr option\n\nThe autodoc_index_modules option deprecated in pbr 4.2.\n\nChange-Id: Iabdf2595fd9645ce66f3c9377e48074cb71e4cc8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e6006cc1ad1f5c85b70c49df4ea6fa4a3f073100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e6006cc1ad1f5c85b70c49df4ea6fa4a3f073100"}]},"branch":"refs/heads/master"},"2e31214c54b047f0945f467f590091bae680fa33":{"kind":"REWORK","_number":2,"created":"2025-07-22 10:47:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/955580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/80/955580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/80/955580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/80/955580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/80/955580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/80/955580/2"}}},"commit":{"parents":[{"commit":"d54f6c655a8738f25cdc70cb0a323f18995f4884","subject":"Remove default override for config options policy_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/d54f6c655a8738f25cdc70cb0a323f18995f4884"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 10:46:38.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 10:47:18.000000000","tz":0},"subject":"Remove deprecated pbr option","message":"Remove deprecated pbr option\n\nThe autodoc_index_modules option deprecated in pbr 4.2.\n\nChange-Id: Iabdf2595fd9645ce66f3c9377e48074cb71e4cc8\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2e31214c54b047f0945f467f590091bae680fa33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2e31214c54b047f0945f467f590091bae680fa33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
