)]}'
{"id":"openstack%2Fmonasca-common~955677","triplet_id":"openstack%2Fmonasca-common~master~Ia0b4308ebe97a7b008f3795710f955cb31b8b8f2","project":"openstack/monasca-common","branch":"master","attention_set":{},"removed_from_attention_set":{"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2025-07-23 10:35:21.000000000","reason":"\u003cGERRIT_ACCOUNT_13114\u003e replied on the change","reason_account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-23 11:53:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0b4308ebe97a7b008f3795710f955cb31b8b8f2","subject":"Remove unused file test-blacklist-py3.txt","status":"MERGED","created":"2025-07-23 09:38:31.000000000","updated":"2025-07-23 11:53:07.000000000","submitted":"2025-07-23 11:53:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955677","meta_rev_id":"884ff0b04625f5072f41b3e4d77dc42150565631","_number":955677,"virtual_id_number":955677,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-23 11:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":2,"date":"2025-07-23 10:35:21.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"all":[{"value":1,"date":"2025-07-23 10:35:21.000000000","_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 10:25:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-23 10:28:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"state":"REVIEWER"}],"messages":[{"id":"cc6cb540ca9eb30d2e1c5e752e1b69d89c0e024d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-23 09:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"987253f81a0eef19defb690789a4bcc9c40aa566","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-23 09:39:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8793e5d9deffe5f983912c958194af3e53f0303f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 10:25:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/393ec1f9ed264487b18b740a769b0e09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1df05a6f3e6e47b6ade8b7ef8edb7205 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd90d561d1d40c5aac134d7a043b3e3 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/452c52593064449c8f3de3d5db0d2ba2 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a5fb229995f4cd893f969a88abe94cf : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b05a5a30f6804e4eaeafea6e2593a0ef : SUCCESS in 7m 26s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/c5c5fc33e48d49f7b4cecf720154c4a2 : FAILURE in 40m 24s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/21d7724455c1489fb22f313127dd3e2d : FAILURE in 43m 23s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/ddd38afa123c4f94a0c0967621574074 : SUCCESS in 26s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/8e583b37f3744ae3b6f58c4479e8b15d : SUCCESS in 1m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7e1b384ed4c46d043d3947b639f6bfdf5d2ba29","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-23 10:28:07.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"63d4dcf829880a8ed6f3cbdb6cd387e6eaeebcf7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-23 10:28:10.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"03d8385d42a2231c7294eb539806b6d37b52cee6","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2025-07-23 10:35:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ba5138085390b1a0430cf119f62f369a1d714007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 11:24:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/763d3ef3ccc34badaf9a335ed10d4497\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47632a9315b64be39d85bd84ba35664c : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c716723ffc6b4dd39f4f706453811134 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dce6331faba64a3fb546bbd5bb704e38 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b1f7ba52b5d4ad9b04eae15fc6b29a3 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56d4db2199744f4695bcc4d1b1f0fdff : SUCCESS in 3m 17s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/39977f5a978e46f0b1a255eb7fb07335 : FAILURE in 32m 55s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/718aa7cdb29d479a9aeac23a66660b38 : FAILURE in 32m 49s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/ecd989c9d5f141ed8a5894f2bf9db15e : SUCCESS in 28s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/c980bb21ef1d46a491d542015cc916f1 : SUCCESS in 1m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9519319b0480af807f66743ec2cee2790ab759dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 11:24:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5438d15066592b7c3b6c119e0dda3894fef772ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 11:53:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a77604822e045d4833a26efc733441f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5329487e143e4805a1063906936a832b : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcc307ec0e53411a995448ff2087bda0 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1490382ca8b64ed8a5964b4ccab7ce3a : SUCCESS in 5m 44s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/9d83c0e674504f0693f305119ff20303 : FAILURE in 27m 01s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/baa9fa9c399b4bdcbd21616f57e12e81 : SUCCESS in 25s","accounts_in_message":[],"_revision_number":3},{"id":"884ff0b04625f5072f41b3e4d77dc42150565631","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 11:53:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4d0d29a675f361e22afe14651aff8ddb87be6e16","revisions":{"aa16860811a8e30e3014e76a575b6d87817b62c7":{"kind":"REWORK","_number":1,"created":"2025-07-23 09:38:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/77/955677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/77/955677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/1"}}},"commit":{"parents":[{"commit":"afa17baad93e204bc2bcce8c886b4df108eacd20","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/afa17baad93e204bc2bcce8c886b4df108eacd20"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-23 09:38:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-23 09:38:31.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ia0b4308ebe97a7b008f3795710f955cb31b8b8f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/aa16860811a8e30e3014e76a575b6d87817b62c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/aa16860811a8e30e3014e76a575b6d87817b62c7"}]},"branch":"refs/heads/master"},"dc32d3f15ad0adfbd4591dfb5a0e85c0d92dc8ed":{"kind":"REWORK","_number":2,"created":"2025-07-23 09:39:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/77/955677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/77/955677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/2"}}},"commit":{"parents":[{"commit":"afa17baad93e204bc2bcce8c886b4df108eacd20","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/afa17baad93e204bc2bcce8c886b4df108eacd20"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-23 09:38:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-23 09:38:58.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ia0b4308ebe97a7b008f3795710f955cb31b8b8f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/dc32d3f15ad0adfbd4591dfb5a0e85c0d92dc8ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/dc32d3f15ad0adfbd4591dfb5a0e85c0d92dc8ed"}]},"branch":"refs/heads/master"},"4d0d29a675f361e22afe14651aff8ddb87be6e16":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-07-23 10:28:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/77/955677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/77/955677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/77/955677/3"}}},"commit":{"parents":[{"commit":"afa17baad93e204bc2bcce8c886b4df108eacd20","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/afa17baad93e204bc2bcce8c886b4df108eacd20"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-23 09:38:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-23 10:28:07.000000000","tz":0},"subject":"Remove unused file test-blacklist-py3.txt","message":"Remove unused file test-blacklist-py3.txt\n\nChange-Id: Ia0b4308ebe97a7b008f3795710f955cb31b8b8f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/4d0d29a675f361e22afe14651aff8ddb87be6e16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/4d0d29a675f361e22afe14651aff8ddb87be6e16"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
