)]}'
{"id":"openstack%2Fmonasca-events-api~669260","triplet_id":"openstack%2Fmonasca-events-api~master~I34255a603c069ed3b1e52017634ff01147473221","project":"openstack/monasca-events-api","branch":"master","topic":"broken_unit_tests_falcon","hashtags":[],"change_id":"I34255a603c069ed3b1e52017634ff01147473221","subject":"Add support for using Falcon 2.0.0","status":"MERGED","created":"2019-07-05 06:30:17.000000000","updated":"2019-07-12 13:40:38.000000000","submitted":"2019-07-12 13:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669260-1562938838929-0ae53db0","meta_rev_id":"27e30c6d269d02be1a1717ae9c58f315a21b240e","_number":669260,"virtual_id_number":669260,"owner":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-12 08:35:59.000000000","_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":2,"date":"2019-07-12 13:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-12 12:34:32.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":2,"date":"2019-07-12 11:18:19.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 12:34:32.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 11:18:19.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-07-12 12:34:32.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-07-12 13:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b21489fc6c8aba35a76c1c22bc364d19ddfb678d","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-05 06:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"672ca35e5290120898f68cc6e186c899d3c75203","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-05 06:38:24.000000000","message":"Topic set to broken_unit_tests_falcon","accounts_in_message":[],"_revision_number":1},{"id":"35d89136a825c26ce27d197fa754d0d7f770e5b7","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-05 06:42:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7b08f3c5f966e52aabeb8cb979d8ca31b1186e7a","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-05 07:34:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98649ad638013060b9e6c9311a7f8498d262c28d","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-05 07:36:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4242fde9ec9214b8a739516a7be1d1e5eab52a61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 08:40:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/60/669260/3/check/build-openstack-api-ref/40d2a02/html/ : SUCCESS in 3m 03s\n- openstack-tox-cover http://logs.openstack.org/60/669260/3/check/openstack-tox-cover/1bb94c7/cover/ : SUCCESS in 10m 11s\n- openstack-tox-pep8 http://logs.openstack.org/60/669260/3/check/openstack-tox-pep8/47a0405/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/60/669260/3/check/openstack-tox-py27/67d70ad/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/60/669260/3/check/openstack-tox-py36/8b1baf7/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/60/669260/3/check/openstack-tox-py37/acf70ea/ : SUCCESS in 3m 16s\n- requirements-check http://logs.openstack.org/60/669260/3/check/requirements-check/2b4b0d6/ : FAILURE in 2m 26s\n- openstack-tox-docs http://logs.openstack.org/60/669260/3/check/openstack-tox-docs/30394c6/html/ : SUCCESS in 3m 14s\n- build-openstack-releasenotes http://logs.openstack.org/60/669260/3/check/build-openstack-releasenotes/abc3c1b/html/ : SUCCESS in 2m 55s\n- monasca-events-python2 http://logs.openstack.org/60/669260/3/check/monasca-events-python2/f968fe4/ : SUCCESS in 56m 24s\n- monasca-events-python3 http://logs.openstack.org/60/669260/3/check/monasca-events-python3/818dd9c/ : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":3},{"id":"91d4e5d681a7c05da867cbfbec9d7a381f51e07d","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-05 10:12:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d1b43bfb2b768e3c0f480485c778d3f622d6115b","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-05 11:16:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a587dc8aeb1f545b6aff9a2ca61dac0ac3283009","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-05 11:23:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"81754ebadf81aeb1262727d10af6d0e7803eb832","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-05 11:37:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c40b25ba271c3ca2d8d832ea96520150153a7e2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 12:42:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/60/669260/6/check/build-openstack-api-ref/67cc29a/html/ : SUCCESS in 2m 59s\n- openstack-tox-cover http://logs.openstack.org/60/669260/6/check/openstack-tox-cover/12c9387/cover/ : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/669260/6/check/openstack-tox-lower-constraints/9d8fb38/ : FAILURE in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/60/669260/6/check/openstack-tox-pep8/6cb8adc/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/60/669260/6/check/openstack-tox-py27/1e114e2/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/60/669260/6/check/openstack-tox-py36/e6623a2/ : SUCCESS in 2m 52s\n- openstack-tox-py37 http://logs.openstack.org/60/669260/6/check/openstack-tox-py37/270dfc1/ : SUCCESS in 5m 54s\n- requirements-check http://logs.openstack.org/60/669260/6/check/requirements-check/b553954/ : SUCCESS in 2m 01s\n- openstack-tox-docs http://logs.openstack.org/60/669260/6/check/openstack-tox-docs/594ea51/html/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/60/669260/6/check/build-openstack-releasenotes/f56ffe3/html/ : SUCCESS in 3m 07s\n- monasca-events-python2 http://logs.openstack.org/60/669260/6/check/monasca-events-python2/fe82393/ : SUCCESS in 47m 56s\n- monasca-events-python3 http://logs.openstack.org/60/669260/6/check/monasca-events-python3/5792f1b/ : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":6},{"id":"8043a51e7d375791e537b226fe2585e5d973b621","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-09 10:23:13.000000000","message":"Patch Set 6:\n\nI think lower-costraints job is failing because lower-constraints.txt file is missing.\n\nhttp://logs.openstack.org/60/669260/6/check/openstack-tox-lower-constraints/9d8fb38/job-output.txt.gz#_2019-07-05_11_46_02_513250","accounts_in_message":[],"_revision_number":6},{"id":"78f20599958376f7bca5ecaf75bb3e9a2086f5ff","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-11 04:40:11.000000000","message":"Patch Set 6:\n\n\u003e I think lower-costraints job is failing because lower-constraints.txt\n \u003e file is missing.\n \u003e \n \u003e http://logs.openstack.org/60/669260/6/check/openstack-tox-lower-constraints/9d8fb38/job-output.txt.gz#_2019-07-05_11_46_02_513250\n\nThanks Adrian for reviewing. But I am adding the file in this patch. Please tell me if I am missing something.","accounts_in_message":[],"_revision_number":6},{"id":"777ac7c84d712aa15852112107293adc026f1bad","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-11 08:49:34.000000000","message":"Patch Set 6:\n\n\u003e \u003e I think lower-costraints job is failing because lower-constraints.txt\n \u003e \u003e file is missing.\n \u003e \u003e\n \u003e \u003e http://logs.openstack.org/60/669260/6/check/openstack-tox-lower-constraints/9d8fb38/job-output.txt.gz#_2019-07-05_11_46_02_513250\n \u003e \n \u003e Thanks Adrian for reviewing. But I am adding the file in this\n \u003e patch. Please tell me if I am missing something.\n\nEverything looks okay to me, but i think there is problem with simplejson library. You can try to use the same version that we are using in monasca-log-api  https://github.com/openstack/monasca-log-api/blob/master/lower-constraints.txt#L77","accounts_in_message":[],"_revision_number":6},{"id":"c3e30b5c7e4f15ee67d4b517f05a63cd85cf7758","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-12 06:39:23.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"636f0394471b8d2147edf4e9806f8bbfaa982808","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 07:39:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/60/669260/7/check/build-openstack-api-ref/dbbfa24/html/ : SUCCESS in 2m 27s\n- openstack-tox-cover http://logs.openstack.org/60/669260/7/check/openstack-tox-cover/003cb22/cover/ : SUCCESS in 3m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/669260/7/check/openstack-tox-lower-constraints/7d13693/ : SUCCESS in 2m 54s\n- openstack-tox-pep8 http://logs.openstack.org/60/669260/7/check/openstack-tox-pep8/ea8b453/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/60/669260/7/check/openstack-tox-py27/cbf42f6/ : SUCCESS in 5m 30s\n- openstack-tox-py36 http://logs.openstack.org/60/669260/7/check/openstack-tox-py36/aaa8ace/ : SUCCESS in 6m 26s\n- openstack-tox-py37 http://logs.openstack.org/60/669260/7/check/openstack-tox-py37/588241b/ : SUCCESS in 6m 34s\n- requirements-check http://logs.openstack.org/60/669260/7/check/requirements-check/bd472dc/ : SUCCESS in 2m 21s\n- openstack-tox-docs http://logs.openstack.org/60/669260/7/check/openstack-tox-docs/bcc1437/html/ : SUCCESS in 3m 01s\n- build-openstack-releasenotes http://logs.openstack.org/60/669260/7/check/build-openstack-releasenotes/b79a94e/html/ : SUCCESS in 2m 55s\n- monasca-events-python2 http://logs.openstack.org/60/669260/7/check/monasca-events-python2/4e4daf6/ : SUCCESS in 48m 14s\n- monasca-events-python3 http://logs.openstack.org/60/669260/7/check/monasca-events-python3/f149f02/ : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":7},{"id":"752d10cb250cda84996af95248aca7cf15277483","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-07-12 08:35:59.000000000","message":"Patch Set 7:\n\n\u003e \u003e \u003e I think lower-costraints job is failing because\n \u003e lower-constraints.txt\n \u003e \u003e \u003e file is missing.\n \u003e \u003e \u003e\n \u003e \u003e \u003e http://logs.openstack.org/60/669260/6/check/openstack-tox-lower-constraints/9d8fb38/job-output.txt.gz#_2019-07-05_11_46_02_513250\n \u003e \u003e\n \u003e \u003e Thanks Adrian for reviewing. But I am adding the file in this\n \u003e \u003e patch. Please tell me if I am missing something.\n \u003e \n \u003e Everything looks okay to me, but i think there is problem with\n \u003e simplejson library. You can try to use the same version that we are\n \u003e using in monasca-log-api  https://github.com/openstack/monasca-log-api/blob/master/lower-constraints.txt#L77\n\nThanks Adrian, it helped","accounts_in_message":[],"_revision_number":7},{"id":"af29d395a8837833d13833c9c208b924224c2253","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-12 11:18:19.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks.","accounts_in_message":[],"_revision_number":7},{"id":"3a4f26dde8bd8e8717217e4dfe05a1d55691d5f9","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-12 12:34:32.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1fd9a2cd6a284a99e5ecf8a80b7d5f9980fb803d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 12:34:45.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e03ada755f971e7402642ce008c61ecf9eea2065","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 13:40:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/60/669260/7/gate/build-openstack-api-ref/1ab08a2/html/ : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/669260/7/gate/openstack-tox-lower-constraints/611bc45/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/60/669260/7/gate/openstack-tox-pep8/2df2fff/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/60/669260/7/gate/openstack-tox-py27/13549c8/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/60/669260/7/gate/openstack-tox-py36/b8afb1e/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/60/669260/7/gate/openstack-tox-py37/c3b551d/ : SUCCESS in 3m 46s\n- requirements-check http://logs.openstack.org/60/669260/7/gate/requirements-check/4f8012c/ : SUCCESS in 2m 24s\n- openstack-tox-docs http://logs.openstack.org/60/669260/7/gate/openstack-tox-docs/d8c1dcb/html/ : SUCCESS in 3m 29s\n- build-openstack-releasenotes http://logs.openstack.org/60/669260/7/gate/build-openstack-releasenotes/69862be/html/ : SUCCESS in 3m 14s\n- monasca-events-python2 http://logs.openstack.org/60/669260/7/gate/monasca-events-python2/8719d64/ : SUCCESS in 58m 05s\n- monasca-events-python3 http://logs.openstack.org/60/669260/7/gate/monasca-events-python3/3bc540b/ : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":7},{"id":"27e30c6d269d02be1a1717ae9c58f315a21b240e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 13:40:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8ec3b67b77c76db1f3a001450a246fc6d8a8047f","revisions":{"6d53b657edee49609cadf3943a1900f977a561f0":{"kind":"REWORK","_number":1,"created":"2019-07-05 06:30:17.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/60/669260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/60/669260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/1"}}},"commit":{"parents":[{"commit":"6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:26:47.000000000","tz":330},"committer":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:30:13.000000000","tz":330},"subject":"Add support for using Falcon 2.0.0","message":"Add support for using Falcon 2.0.0\n\nFalcon 2.0.0 introduces some breaking changes. The relevant ones here are:\n\n- falcon.testing.TestCase.api property was removed\n- falcon.testing.TestBase class was removed\n\nChange-Id: I34255a603c069ed3b1e52017634ff01147473221\nStory: 2005695\nTask: 35688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6d53b657edee49609cadf3943a1900f977a561f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6d53b657edee49609cadf3943a1900f977a561f0"}]},"branch":"refs/heads/master"},"966bb5a7bd7f0f675afa893b819c34217920d3e9":{"kind":"REWORK","_number":2,"created":"2019-07-05 06:42:45.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/60/669260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/60/669260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/2"}}},"commit":{"parents":[{"commit":"6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:26:47.000000000","tz":330},"committer":{"name":"Akhil jain","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:41:36.000000000","tz":0},"subject":"Add support for using Falcon 2.0.0","message":"Add support for using Falcon 2.0.0\n\nFalcon 2.0.0 introduces some breaking changes. The relevant ones here are:\n\n- falcon.testing.TestCase.api property was removed\n- falcon.testing.TestBase class was removed\n\nChange-Id: I34255a603c069ed3b1e52017634ff01147473221\nStory: 2005695\nTask: 35688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/966bb5a7bd7f0f675afa893b819c34217920d3e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/966bb5a7bd7f0f675afa893b819c34217920d3e9"}]},"branch":"refs/heads/master"},"e1bb6f0f1042206360d7d9d3f87e8dbdf8fd4a92":{"kind":"REWORK","_number":3,"created":"2019-07-05 07:34:58.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/60/669260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/60/669260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/3"}}},"commit":{"parents":[{"commit":"6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:26:47.000000000","tz":330},"committer":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 07:33:05.000000000","tz":330},"subject":"Add support for using Falcon 2.0.0","message":"Add support for using Falcon 2.0.0\n\nFalcon 2.0.0 introduces some breaking changes. The relevant ones here are:\n\n- falcon.testing.TestCase.api property was removed\n- falcon.testing.TestBase class was removed\n\nThis commit also switch project use stestr and update sphinx dependency\n\nChange-Id: I34255a603c069ed3b1e52017634ff01147473221\nStory: 2005695\nTask: 35688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/e1bb6f0f1042206360d7d9d3f87e8dbdf8fd4a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/e1bb6f0f1042206360d7d9d3f87e8dbdf8fd4a92"}]},"branch":"refs/heads/master"},"cebb4b721e1f8ac881b2819c9cc9d2574ecd92fd":{"kind":"REWORK","_number":4,"created":"2019-07-05 10:12:50.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/60/669260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/60/669260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/4"}}},"commit":{"parents":[{"commit":"6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:26:47.000000000","tz":330},"committer":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 10:12:07.000000000","tz":330},"subject":"Add support for using Falcon 2.0.0","message":"Add support for using Falcon 2.0.0\n\nFalcon 2.0.0 introduces some breaking changes. The relevant ones here are:\n\n- falcon.testing.TestCase.api property was removed\n- falcon.testing.TestBase class was removed\n\nThis commit also switches the project to use stestr, update sphinx dependency\nand add lower-constraints job\n\nChange-Id: I34255a603c069ed3b1e52017634ff01147473221\nStory: 2005695\nTask: 35688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/cebb4b721e1f8ac881b2819c9cc9d2574ecd92fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/cebb4b721e1f8ac881b2819c9cc9d2574ecd92fd"}]},"branch":"refs/heads/master"},"74b95b6a6efd838126e4ac8d345e17a8548ede2d":{"kind":"REWORK","_number":5,"created":"2019-07-05 11:16:29.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/60/669260/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/60/669260/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/5"}}},"commit":{"parents":[{"commit":"6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:26:47.000000000","tz":330},"committer":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 11:16:09.000000000","tz":330},"subject":"Add support for using Falcon 2.0.0","message":"Add support for using Falcon 2.0.0\n\nFalcon 2.0.0 introduces some breaking changes. The relevant ones here are:\n\n- falcon.testing.TestCase.api property was removed\n- falcon.testing.TestBase class was removed\n\nThis commit also switches the project to use stestr, update sphinx dependency\nand add lower-constraints job\n\nChange-Id: I34255a603c069ed3b1e52017634ff01147473221\nStory: 2005695\nTask: 35688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/74b95b6a6efd838126e4ac8d345e17a8548ede2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/74b95b6a6efd838126e4ac8d345e17a8548ede2d"}]},"branch":"refs/heads/master"},"d5e7af909ac10a76e37c972da1f6641f7ff1c8ac":{"kind":"REWORK","_number":6,"created":"2019-07-05 11:37:43.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/60/669260/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/60/669260/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/6"}}},"commit":{"parents":[{"commit":"6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:26:47.000000000","tz":330},"committer":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 11:37:26.000000000","tz":330},"subject":"Add support for using Falcon 2.0.0","message":"Add support for using Falcon 2.0.0\n\nFalcon 2.0.0 introduces some breaking changes. The relevant ones here are:\n\n- falcon.testing.TestCase.api property was removed\n- falcon.testing.TestBase class was removed\n\nThis commit also switches the project to use stestr, update sphinx dependency\nand add lower-constraints job\n\nChange-Id: I34255a603c069ed3b1e52017634ff01147473221\nStory: 2005695\nTask: 35688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/d5e7af909ac10a76e37c972da1f6641f7ff1c8ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/d5e7af909ac10a76e37c972da1f6641f7ff1c8ac"}]},"branch":"refs/heads/master"},"8ec3b67b77c76db1f3a001450a246fc6d8a8047f":{"kind":"REWORK","_number":7,"created":"2019-07-12 06:39:23.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/60/669260/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/60/669260/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/60/669260/7"}}},"commit":{"parents":[{"commit":"6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6647b88cbe78d6fc0fabe4ef332ca21a3e74fe3a"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2019-07-05 06:26:47.000000000","tz":330},"committer":{"name":"Akhil jain","email":"akhiljain32@gmail.com","date":"2019-07-12 06:39:15.000000000","tz":0},"subject":"Add support for using Falcon 2.0.0","message":"Add support for using Falcon 2.0.0\n\nFalcon 2.0.0 introduces some breaking changes. The relevant ones here are:\n\n- falcon.testing.TestCase.api property was removed\n- falcon.testing.TestBase class was removed\n\nThis commit also switches the project to use stestr, update sphinx dependency\nand add lower-constraints job\n\nChange-Id: I34255a603c069ed3b1e52017634ff01147473221\nStory: 2005695\nTask: 35688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/8ec3b67b77c76db1f3a001450a246fc6d8a8047f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/8ec3b67b77c76db1f3a001450a246fc6d8a8047f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
