)]}'
{"id":"openstack%2Fmonasca-events-api~689677","triplet_id":"openstack%2Fmonasca-events-api~master~Ia865a8b26da082a7227d2afce82e3c164acb8efb","project":"openstack/monasca-events-api","branch":"master","topic":"fix-admin-policy","hashtags":[],"change_id":"Ia865a8b26da082a7227d2afce82e3c164acb8efb","subject":"Fixes policy for admin","status":"ABANDONED","created":"2019-10-21 06:33:36.000000000","updated":"2026-02-03 22:33:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d7271fafc99ecaa8b83aaa971d1e3df1328b1eba","_number":689677,"virtual_id_number":689677,"owner":{"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-02 10:41:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-02 09:44:53.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 09:44:53.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-12-02 10:41:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 22:33:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"b270044e853cf9b032d3505e5396b0ad538d5879","author":{"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},"date":"2019-10-21 06:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70007e3641367bfb4040a29e88d27df629b2eed2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 07:39:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de0259e70c074fafbb6cb9f63134287e : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94b6908181ca4ee88d2193a9feeaabca : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf32d742322490384981fb2e298e970 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f546b6a8f6f74ee0b2d37ad553c38bf1 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0630dde8a464e14b59d4d648785175b : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53010a7d2b874e2bbdfebc1ae92962ec : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/519364ec5b754096b5039ea59a4d0b3f : FAILURE in 4m 18s\n- monasca-events-python2 https://zuul.opendev.org/t/openstack/build/c4b0fde2aa9b4aa7bcd9b02be17c4737 : FAILURE in 1h 00m 23s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/d704e914a9664bfe8baa79f303826879 : FAILURE in 1h 03m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ab2306c3818676eb652a8229b4e2a57669485daf","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-12-02 09:44:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b497011934ae5e960d51018648226d0aa2cc981a","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-12-02 09:44:53.000000000","message":"Patch Set 2:\n\nDo we need this rule defined here at all?","accounts_in_message":[],"_revision_number":2},{"id":"082c8f570c16e00d103764946a0bacca0dcb3ec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 10:41:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83ba1b7636fc45d8826da9b01f6e5b42 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca2f6f868da64cb68bd0a7f9d5ae1ee3 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b21d1df39234b4a804d2d6061d14646 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2386124c9d544e658a082c9b0e781623 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba69ec3b0bb420dac6bcc7ae8020212 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc6fd24060c24b9b9e237e762e85097e : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad2fb75f62940869f7f46e7024b4ce7 : SUCCESS in 3m 28s\n- monasca-events-python2 https://zuul.opendev.org/t/openstack/build/1b773ca265c14f05a78ed32fb661b099 : SUCCESS in 53m 17s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/0a6a20e676ea4611b0982d67616b0913 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d7271fafc99ecaa8b83aaa971d1e3df1328b1eba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:33:23.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5050a0ec7c80e58fa0d7f651cd968d53163659f6","revisions":{"0748b630ac8b08f83048f68fe05830dadb1cd0bc":{"kind":"REWORK","_number":1,"created":"2019-10-21 06:33:36.000000000","uploader":{"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},"ref":"refs/changes/77/689677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/77/689677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/689677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/689677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/689677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/77/689677/1"}}},"commit":{"parents":[{"commit":"64e4e9af47c0a4b62f5c21629c20ed1b12aea0fc","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/64e4e9af47c0a4b62f5c21629c20ed1b12aea0fc"}]}],"author":{"name":"liyingjun","email":"yingjun.li@kylin-cloud.com","date":"2019-10-21 06:28:34.000000000","tz":480},"committer":{"name":"liyingjun","email":"yingjun.li@kylin-cloud.com","date":"2019-10-21 06:33:29.000000000","tz":480},"subject":"Fixes policy for admin","message":"Fixes policy for admin\n\nChange-Id: Ia865a8b26da082a7227d2afce82e3c164acb8efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/0748b630ac8b08f83048f68fe05830dadb1cd0bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/0748b630ac8b08f83048f68fe05830dadb1cd0bc"}]},"branch":"refs/heads/master"},"5050a0ec7c80e58fa0d7f651cd968d53163659f6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-02 09:44:24.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/77/689677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/77/689677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/689677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/689677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/689677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/77/689677/2"}}},"commit":{"parents":[{"commit":"5e7f9f193d8eb264554479f816b7aea906fa0aad","subject":"Bump the openstackdocstheme extension to 1.20.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/5e7f9f193d8eb264554479f816b7aea906fa0aad"}]}],"author":{"name":"liyingjun","email":"yingjun.li@kylin-cloud.com","date":"2019-10-21 06:28:34.000000000","tz":480},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@suse.com","date":"2019-12-02 09:44:24.000000000","tz":0},"subject":"Fixes policy for admin","message":"Fixes policy for admin\n\nChange-Id: Ia865a8b26da082a7227d2afce82e3c164acb8efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/5050a0ec7c80e58fa0d7f651cd968d53163659f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/5050a0ec7c80e58fa0d7f651cd968d53163659f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
