)]}'
{"id":"openstack%2Fmonasca-events-api~717553","triplet_id":"openstack%2Fmonasca-events-api~master~I42a8071625d8e309d1580f31e440df4d54d2d37b","project":"openstack/monasca-events-api","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"I42a8071625d8e309d1580f31e440df4d54d2d37b","subject":"Cleanup py27 support","status":"ABANDONED","created":"2020-04-05 07:53:56.000000000","updated":"2021-08-25 15:04:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1737fd2b751cbdf0d3721251133cc08e76c0f030","_number":717553,"virtual_id_number":717553,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-22 06:11:32.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-05 09:05:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-05 09:05:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-07 13:13:52.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"221d232ec6cb2d2712bd590783b5d023945b3148","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-05 07:53:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abc0e576f0e72503f717c088500086b0787eb52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 08:09:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/af722bc80e41461082c1c538d6d9ce7f : SUCCESS in 3m 39s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c98f9330851417bb259ba951b59d689 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a407b25d311465194283c5dbb722eca : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71046b50464f479abb6fa66765060931 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d473d3b0b648a8a22a26baecf61e35 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/410a3dced0514be292501f5814b66b8b : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f1964340d974f06a0dfe361fba3864c : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ae88c5acbc47559a6a0b5b002fede5 : SUCCESS in 3m 35s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/c154700e2e024f3e983d7865acc56f9a : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3928d01fb5cc4735a6c5ec069e90095e : FAILURE in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fb167cf91424e379e3b413137dc504a : SUCCESS in 4m 13s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/919bafa9e8d54f5a8a2a3229e664fefc : FAILURE in 13m 29s\n\nWarning:\n  Comments left for invalid file doc/source/contributor/code.rst","accounts_in_message":[],"_revision_number":1},{"id":"65ecf7c73ba04c9ab7b180e50c05fab85c9b320b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-05 08:46:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce6888da7ea58a848416bc78be0f7f118c3318f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 09:05:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4dc70c49f10e485cb8c0f91320a49300 : SUCCESS in 3m 40s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21db97f1364143c492da6d6d8393d5be : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/742acd783c344d66bf629f183db42113 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df1ec8ab16764558b981e4cf4376651e : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18a2f7b4ff7450ca2138101b9108645 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f9959bf05c045e6bf5bb96dde49ab99 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6eb47b136df344728cbe66f399df8b1a : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c993344d5a449aca6bcfd5a73809500 : SUCCESS in 3m 35s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/655a621ef67c471ea7c2454e57133051 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102c258f96b54b02b9dae75b0698e7e9 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6183a5a12184e6a950e3cdcb4a4842d : SUCCESS in 3m 28s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/efb4d3638738414694103c9d50ccb665 : FAILURE in 13m 46s\n\nWarning:\n  Comments left for invalid file doc/source/contributor/code.rst","accounts_in_message":[],"_revision_number":2},{"id":"55828cef88efb9108ddcfdc826f5103f799a326b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 06:11:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"92fdb84a5827c55a3131dbcceb64f1ef91f9034d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 07:16:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/778c3fd331d44ca2a38b7c42c7deca50 : SUCCESS in 3m 39s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/309c72f0f06446638fb5a0e1f9ab8f5d : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/902bef0608fa4ee79f3d056d1d663adc : FAILURE in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c411b8b88c0c485fb0b56d305a0c9bce : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a556843dea734446a9db46a3ab4d2c47 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eb95bc7b1fd418caca7b8cccf8b519e : FAILURE in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d747a81af134e7bad3bced976f819f9 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd128d7515b4bdea2837524e2cf12aa : SUCCESS in 3m 58s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/d791c195d3d14cedb2cb01ed7ab004d7 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e515e8664a374718a678163ee20e8ffe : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a784ab42a2084b169c5e879dc8dc083e : SUCCESS in 3m 18s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/abeb1064653a4b158aca23c4a845bbf2 : FAILURE in 15m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1737fd2b751cbdf0d3721251133cc08e76c0f030","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-08-25 15:04:23.000000000","message":"Abandoned\n\nNo interest from team to take this, I won\u0027t push further for this.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"11a69eb678bd5be6abbb5ba2ef5ce4e6392e7500","revisions":{"f572bfbabd216d8227d402b49bb843a614dbbc89":{"kind":"REWORK","_number":1,"created":"2020-04-05 07:53:56.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/53/717553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/53/717553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/53/717553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/53/717553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/53/717553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/53/717553/1"}}},"commit":{"parents":[{"commit":"2550100ada77785e0af8e85f08afb806a6383e0a","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/2550100ada77785e0af8e85f08afb806a6383e0a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 07:51:57.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 07:53:37.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg:\n  * Wheel is not needed for python 3 only repo\n  * Some other sections are obsolete\n- Update classifiers\n- Update requirements, no need for python_version anymore\n- Use sphinx-build everywhere\n\nChange-Id: I42a8071625d8e309d1580f31e440df4d54d2d37b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f572bfbabd216d8227d402b49bb843a614dbbc89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f572bfbabd216d8227d402b49bb843a614dbbc89"}]},"branch":"refs/heads/master"},"11a69eb678bd5be6abbb5ba2ef5ce4e6392e7500":{"kind":"REWORK","_number":2,"created":"2020-04-05 08:46:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/53/717553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/53/717553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/53/717553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/53/717553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/53/717553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/53/717553/2"}}},"commit":{"parents":[{"commit":"2550100ada77785e0af8e85f08afb806a6383e0a","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/2550100ada77785e0af8e85f08afb806a6383e0a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 07:51:57.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 08:46:05.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg:\n  * Wheel is not needed for python 3 only repo\n  * Some other sections are obsolete\n- Update classifiers\n- Update requirements, no need for python_version anymore\n- Use sphinx-build everywhere\n\nChange-Id: I42a8071625d8e309d1580f31e440df4d54d2d37b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/11a69eb678bd5be6abbb5ba2ef5ce4e6392e7500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/11a69eb678bd5be6abbb5ba2ef5ce4e6392e7500"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
