)]}'
{"id":"openstack%2Fmonasca-events-api~720942","triplet_id":"openstack%2Fmonasca-events-api~master~I91fdc1d44a250f72de58166420835df939152b51","project":"openstack/monasca-events-api","branch":"master","topic":"unittest.mock","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-02-02 15:02:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I91fdc1d44a250f72de58166420835df939152b51","subject":"Use unittest.mock instead of third party mock","status":"ABANDONED","created":"2020-04-18 16:55:03.000000000","updated":"2026-02-02 15:02:51.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"83d471223d112bd8a90c8e6c9cd5e2e3bf0918da","_number":720942,"virtual_id_number":720942,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-27 22:00:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 10:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67a4530ba4a6a7a335e2fe7a69d09f2bdbc91858","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-18 16:55:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84efda95b836c8ffb70f1107725ecaa97dbe33c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 12:23:47.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/95f315fc144e47b88167c056d6e1eb65 : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b94f170b90fd48089a72ea5ae1abc5bd : FAILURE in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c60f280cfd814f799f3b700b88957625 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee548b4ca044627b1cd082f9dbeb2ca : FAILURE in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4136844353de41e9a128f6ed34a93032 : FAILURE in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb20943cdc5d486eb026c9713037992f : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b40d06d879d54877a017bbb3b36e6610 : SUCCESS in 6m 47s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/186664f04b654cf6b6416097545ee0b6 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8cc2c97e7df48ecae0326e8030f7b2d : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51041d7c4e1b410a83273c975da029ec : SUCCESS in 3m 35s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/11e0bb9b967f497ea5cfbe9cff20ae32 : FAILURE in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8e88bc80773bebb5ca094cf1ba6ea1df73352247","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-13 10:26:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db6b445269bb945e376dda52222f783b35c5c24a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 10:47:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7b2e15fdb8a7478baa94019074a98d9f : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a93544e0b074617b8fbd14d05175014 : FAILURE in 3m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73fb2348830d4072b653787f23ffc6af : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d2b59a533124ca1b50c5f483cbf278d : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de755db952304a10a5a64d8df632becf : FAILURE in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/402148564ff64cf9a31fb6c09b28abbb : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae0822fd7ffa47ddaf85dcee7809ca16 : SUCCESS in 4m 20s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/88c23b7a85674972bb2ab89cbde6a8ca : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae9d2bc1c6f48c69ba989dfbd2abd2d : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77bd3deb77074ae28019c904c126052d : SUCCESS in 3m 35s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/a31f401c5ede4248bbddecdbb541eebf : FAILURE in 15m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c0d1caaa02d6c0d148110b2322b0feffe9730987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-06-13 19:46:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"df432cddb59a2e80068ecb8b5ec0de3ed36beed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 19:54:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e98dfc5bf3004d45b45207d41868e5bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/136252beda244657a77ce039c60973be : FAILURE in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5132800a4a3345efa9b0078e271a22f1 : FAILURE in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b69b5d53f3e47bbb338c605f783a5e7 : FAILURE in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d746dedea0b491893174c66a803a9bd : FAILURE in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36ec95437ba54eb89fc148cbe56f6837 : FAILURE in 1m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2093e55bd03c45b783edd63ac84dbe85 : FAILURE in 2m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e23f1d71bea745b9b0e99e142c5eedb4 : FAILURE in 1m 54s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/f52dfbf83f664a5c96ba4b2beaa49e23 : SUCCESS in 1m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dedacc8e67654422b63d5358637367cd : FAILURE in 2m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05f24541f33343fcbb2459816fe50c12 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":3},{"id":"ebb41ca39dfe7c2fe9fefb80febd5403515d4f5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-27 14:25:20.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a0227bcb83d3a086e48648529a77ac19bca3c9ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 22:00:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d546af841d843b4ba3298d384ff7c20\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec58f082346e4e0ca480815d75379a21 : FAILURE in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7cd2009bdeb4d2a906ed153025a6b24 : FAILURE in 2m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02e4eac2d9164d8882b0b5f7202cf1e5 : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55059d97720d4665b04da89972b0ad3f : FAILURE in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f63da1f89bfb4d3797f9b7876171626e : FAILURE in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f23cb5cb1140498ab260fd484b1fd3fe : FAILURE in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9ed96b9839e491a9449017221a6eca6 : FAILURE in 1m 50s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/b828210df5b14a39a97fd7ebd618d159 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a4b4315ee94453853944c31ab43735 : FAILURE in 2m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9d3c9e9261d4bf896b4648e93495a87 : SUCCESS in 3m 23s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/590ec7fcf18344db82d34368e9b42205 : FAILURE in 26m 28s","accounts_in_message":[],"_revision_number":4},{"id":"83d471223d112bd8a90c8e6c9cd5e2e3bf0918da","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-02 15:02:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f4f33a6f283b184b37cef21e495e7003ab188664","revisions":{"4f5b10e096e04067b264d21ad0b2ead212a92d61":{"kind":"REWORK","_number":1,"created":"2020-04-18 16:55:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/42/720942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/42/720942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/1"}}},"commit":{"parents":[{"commit":"2550100ada77785e0af8e85f08afb806a6383e0a","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/2550100ada77785e0af8e85f08afb806a6383e0a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:54:53.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:54:53.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I91fdc1d44a250f72de58166420835df939152b51\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/4f5b10e096e04067b264d21ad0b2ead212a92d61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/4f5b10e096e04067b264d21ad0b2ead212a92d61"}]},"branch":"refs/heads/master"},"93f4754f0c2fab37c83acd7cff7b1c5f0fc22dd0":{"kind":"REWORK","_number":2,"created":"2020-05-13 10:26:00.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/42/720942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/42/720942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/2"}}},"commit":{"parents":[{"commit":"2550100ada77785e0af8e85f08afb806a6383e0a","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/2550100ada77785e0af8e85f08afb806a6383e0a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:54:53.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-13 10:25:51.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I91fdc1d44a250f72de58166420835df939152b51\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/93f4754f0c2fab37c83acd7cff7b1c5f0fc22dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/93f4754f0c2fab37c83acd7cff7b1c5f0fc22dd0"}]},"branch":"refs/heads/master"},"0199f16f2573ba408d1d7ca17944e6f82b5e9a59":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-06-13 19:46:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/42/720942/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/42/720942/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/3"}}},"commit":{"parents":[{"commit":"04633984a03e901d077b794b7c185a87756595a7","subject":"Fix deprecated zuul queue syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/04633984a03e901d077b794b7c185a87756595a7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:54:53.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2024-06-13 19:46:29.000000000","tz":0},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I91fdc1d44a250f72de58166420835df939152b51\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/0199f16f2573ba408d1d7ca17944e6f82b5e9a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/0199f16f2573ba408d1d7ca17944e6f82b5e9a59"}]},"branch":"refs/heads/master","description":"Rebase"},"f4f33a6f283b184b37cef21e495e7003ab188664":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-27 14:25:20.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/42/720942/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/42/720942/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/42/720942/4"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:54:53.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 14:25:12.000000000","tz":-360},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I91fdc1d44a250f72de58166420835df939152b51\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f4f33a6f283b184b37cef21e495e7003ab188664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f4f33a6f283b184b37cef21e495e7003ab188664"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
