)]}'
{"id":"openstack%2Fmonasca-events-api~732034","triplet_id":"openstack%2Fmonasca-events-api~master~I4b6d35a0e20e0991091ff3d71d4d7186071d4c2f","project":"openstack/monasca-events-api","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I4b6d35a0e20e0991091ff3d71d4d7186071d4c2f","subject":"Switch to newer openstackdocstheme and reno versions","status":"ABANDONED","created":"2020-05-31 17:44:48.000000000","updated":"2021-08-25 15:03:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c57bea66f3b5c4b42d0a14f6799d478859d8a917","_number":732034,"virtual_id_number":732034,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-31 18:00:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-31 18:00:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c67945a1ad1979da10640ccf5da4dec40afd3007","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-31 17:44:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25109ac4c7f71869db228e1973d0c5bacbe30d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 18:00:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2cdb24e6a3b14a27b1f83e5f586280f0 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85a31d6fcb1a476693c081b1e049bc5e : FAILURE in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e877c9ef7104d5391b7f58a2bc3e8af : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bafe3d033f04159b22f89a0da7ef9e6 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8d638dc25a4657a69244850fd15492 : FAILURE in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38dff76db65c4544835e29afaa499479 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ecccb9df42c46a2a03630c7edf7f49f : SUCCESS in 3m 20s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/12833da116e74db09135ecca89599b43 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e1ff568d784ad4b69ed63198faab17 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/718a8bd52e584503b43c6fb4cb5931f4 : SUCCESS in 2m 54s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/5bc1c1248a57422b9f99787095023dd6 : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c57bea66f3b5c4b42d0a14f6799d478859d8a917","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-08-25 15:03:53.000000000","message":"Abandoned\n\nNo interest from team to take this, I won\u0027t push further for this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4332619acfecad0cd50abaffc609737c6d7bb127","revisions":{"4332619acfecad0cd50abaffc609737c6d7bb127":{"kind":"REWORK","_number":1,"created":"2020-05-31 17:44:48.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/732034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/34/732034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/34/732034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/34/732034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/34/732034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/34/732034/1"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-31 17:42:34.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-31 17:44:36.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I4b6d35a0e20e0991091ff3d71d4d7186071d4c2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/4332619acfecad0cd50abaffc609737c6d7bb127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/4332619acfecad0cd50abaffc609737c6d7bb127"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
