)]}'
{"id":"openstack%2Fmonasca-events-api~734377","triplet_id":"openstack%2Fmonasca-events-api~master~I221e432ac508e3303ab805d6a7990b99a606f2e8","project":"openstack/monasca-events-api","branch":"master","topic":"ujson","hashtags":[],"change_id":"I221e432ac508e3303ab805d6a7990b99a606f2e8","subject":"Use unittest.mock instead of mock","status":"ABANDONED","created":"2020-06-08 23:24:39.000000000","updated":"2020-06-17 19:37:43.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a8f0971a8e0423ea54cf8f3e76d79768ed483339","_number":734377,"virtual_id_number":734377,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-11 12:12:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 12:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b292f2d316c773ba1544e3099c56a5786c27f6a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 23:24:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a766279d402addb8453ef4e7151c5e6322162481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 23:50:23.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9cf659f1484b46238438adce97cc751b : SUCCESS in 5m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b627915f7aae4bd581d48fda6465354a : FAILURE in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0591f3b72234baeb0838caaec75d47f : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed7c2016e556423d810678610ae08635 : FAILURE in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8557ec0ff4b54515859a54911cab93a3 : FAILURE in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d69caf19fd34cbdb7f27a3284131657 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba7f55c703d44035b84f8ca65bc9d7a2 : SUCCESS in 4m 44s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f9d826242964bbab6c12ee53a149341 : FAILURE in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131d2ccd2f704377b02ac64a96d260a2 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eac0f3f0bdcd44d0b91469f4a249408e : SUCCESS in 6m 39s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/2fb3359306844174b15a768eb9848e03 : FAILURE in 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0a21ea36ac162ae50f756185d9306071517a8042","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-10 11:15:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b46873410deaf61768ac95aa4f7c1d79d0b76cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 11:30:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7bd69bcd0c1a447a957e9af81ea546a3 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbf2341a470c4b91ac86708fcb82c537 : FAILURE in 3m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2f86843ebab46578a403166c22b785d : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851e655bfbe14e498dfcdfc2a0ae7153 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f948c0c90e448d08feb13e02827cce3 : FAILURE in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da86edb8bf9043938e3d15d23efc129d : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/599be8c16f8b43acab985d5e587a2151 : SUCCESS in 4m 16s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0c3b0d0c45d4187b3f3e8e1b6a6e9c9 : FAILURE in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d7ea68b2c047669bbe65c2a729d747 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9350d33b63a54345a781a1a30e4e6c2d : SUCCESS in 3m 23s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/30e837454fcb46cfae9544ad254e5d52 : FAILURE in 13m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e54db671930a5da463f84d1d73540186c7f82eda","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-11 11:53:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d3daccd20fb443a85962f38032d71789bcda8ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 12:12:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d76e6b8a69dd410ab0e772eac4eb05a2 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06dd2ea7831247c68b66b78c8fd4a5eb : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64595aa9a8cb4a86b1f7e55e1291dff7 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f86d6c0a2ce406b9c1bb8a786f6befd : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffca66787a2e480f817a7c66773b1238 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2455eb5e3324c7fba503d351d2efca4 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b348d54270384be481e5b8aacd7787ea : SUCCESS in 4m 02s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d3c6afd6ee14ff5a98f9fe9f26b8b44 : FAILURE in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c41845e40f86471d9cefe632d6d5b1b4 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d49f687ac4d54346a26b8dae78191d03 : SUCCESS in 3m 29s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/50ee2c18dd7d44248192700c064ed92a : FAILURE in 17m 56s","accounts_in_message":[],"_revision_number":3},{"id":"a493bc91201fdc59bb13e532a57a836053d7e634","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 19:37:43.000000000","message":"Abandoned\n\n720942","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1a2e9632629d6de58971c04a42dd5fe25c7575fd","revisions":{"4f0053a5dacd406d9d629ee473ba32c457b25571":{"kind":"REWORK","_number":1,"created":"2020-06-08 23:24:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/77/734377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/77/734377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/1"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:24:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:24:34.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: I221e432ac508e3303ab805d6a7990b99a606f2e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/4f0053a5dacd406d9d629ee473ba32c457b25571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/4f0053a5dacd406d9d629ee473ba32c457b25571"}]},"branch":"refs/heads/master"},"abade855086f2be2a9c58faa7182851555cfb72b":{"kind":"REWORK","_number":2,"created":"2020-06-10 11:15:22.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/77/734377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/77/734377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/2"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:24:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-10 11:15:19.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: I221e432ac508e3303ab805d6a7990b99a606f2e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/abade855086f2be2a9c58faa7182851555cfb72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/abade855086f2be2a9c58faa7182851555cfb72b"}]},"branch":"refs/heads/master"},"1a2e9632629d6de58971c04a42dd5fe25c7575fd":{"kind":"REWORK","_number":3,"created":"2020-06-11 11:53:07.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/77/734377/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/77/734377/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/77/734377/3"}}},"commit":{"parents":[{"commit":"fb13387fa0b760ff3b9a1ff8fe08bada3c801bc0","subject":"Drop use of ujson","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fb13387fa0b760ff3b9a1ff8fe08bada3c801bc0"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:24:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-11 11:52:47.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: I221e432ac508e3303ab805d6a7990b99a606f2e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/1a2e9632629d6de58971c04a42dd5fe25c7575fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/1a2e9632629d6de58971c04a42dd5fe25c7575fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
