)]}'
{"id":"openstack%2Fmonasca-events-api~752816","triplet_id":"openstack%2Fmonasca-events-api~master~I01dbe628aa3499eed45c65e9af12d6284b65952b","project":"openstack/monasca-events-api","branch":"master","hashtags":[],"change_id":"I01dbe628aa3499eed45c65e9af12d6284b65952b","subject":"bump py37 to py38 in tox.ini","status":"ABANDONED","created":"2020-09-20 13:52:49.000000000","updated":"2020-10-30 06:12:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ffac6e29b1bdba4ae768625eb5f99f7fdfdb52cb","_number":752816,"virtual_id_number":752816,"owner":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-20 14:06:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-20 14:06:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"755d9a7afa7728ed0d461b38691c807d0cf02996","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2020-09-20 13:52:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d095860c08fb80d20fb5d1c06f12fe66cd0b5216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 14:06:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f5071ca8e074411ba5de1e5bc2b4cd49 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35718416220145fcaeb6df021a6df68a : FAILURE in 3m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3344fb4438f147c8874f6f0bee819c85 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0953d0a13824445a29b1485117adc4c : FAILURE in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93fe43dd25c04c16a2e31286e0de5f5b : FAILURE in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ebd5f31e6fa49e98b1fb523593952b9 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/762be23907a4443da9635d51a5039555 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a90fd83ec7f4836ab9c396fbc48a6aa : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51cefae854c6493ca5a7425b6967f256 : SUCCESS in 3m 17s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/ac021c16ddde443cbcf7989e49e9d2ac : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ffac6e29b1bdba4ae768625eb5f99f7fdfdb52cb","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2020-10-30 06:12:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8ae97c661961994bacbf70fa4a7d4b52efc864a","revisions":{"c8ae97c661961994bacbf70fa4a7d4b52efc864a":{"kind":"REWORK","_number":1,"created":"2020-09-20 13:52:49.000000000","uploader":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"ref":"refs/changes/16/752816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/16/752816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/16/752816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/16/752816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/16/752816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/16/752816/1"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"zhangchun","email":"zhangchun@yovole.com","date":"2020-09-20 13:52:33.000000000","tz":0},"committer":{"name":"zhangchun","email":"zhangchun@yovole.com","date":"2020-09-20 13:52:48.000000000","tz":0},"subject":"bump py37 to py38 in tox.ini","message":"bump py37 to py38 in tox.ini\n\nin \u0027victoria\u0027 cycle, we should test py38 by default.\nTrivial change\n\nChange-Id: I01dbe628aa3499eed45c65e9af12d6284b65952b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/c8ae97c661961994bacbf70fa4a7d4b52efc864a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/c8ae97c661961994bacbf70fa4a7d4b52efc864a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
