)]}'
{"id":"openstack%2Fmonasca-events-api~754774","triplet_id":"openstack%2Fmonasca-events-api~master~I0c20ebb46beb27640afc5258d7cabdfbda25c387","project":"openstack/monasca-events-api","branch":"master","topic":"reno-victoria","hashtags":[],"change_id":"I0c20ebb46beb27640afc5258d7cabdfbda25c387","subject":"Update master for stable/victoria","status":"ABANDONED","created":"2020-09-28 15:48:41.000000000","updated":"2026-02-03 22:32:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2784245b9b51cd836eaa6fc13ea92e90a5b69f68","_number":754774,"virtual_id_number":754774,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-29 14:54:55.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-28 16:23:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-28 16:23:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 14:54:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2026-02-03 22:32:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"ce6600c28eb594cf4a3265315dd5dcb05997adba","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-28 15:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acc633697776409cd2e77fc1963378da37659fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 16:23:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a936c8724d054607a34cdfe4a9335846 : FAILURE in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b221f35b812e4c089d728eef5a07479b : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f26d362e125348cd9cebc2707fb70398 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9efbd49414b65922d4425c24befca75f799e952b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-14 16:45:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d12cf2ec94d5b674df60b478c6ced40a3c2873b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 17:04:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8b1cf8a8744b1791c83fa16a49c25f : FAILURE in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b46a6da9fb4f53b636e5f810173b47 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34ace32f6224441595eafa75b92db40d : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cb567fc8352e5289646b66604f11dff36c7b1865","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-29 14:54:55.000000000","message":"Patch Set 1:\n\nrecheck\n\nI\u0027m not convinced that the recheck will help us to untangle this issue...\n\nWeird context issue on pep8","accounts_in_message":[],"_revision_number":1},{"id":"1df31cf4b14ff9825339c00b9ef70a458a040c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 15:05:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5755affeaf34c1c99748691f8cc5472 : FAILURE in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a336a5fbee04251a5274b0897273929 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1c536c892fd4abea08138a4b71a1bdb : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2784245b9b51cd836eaa6fc13ea92e90a5b69f68","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:32:53.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92fdf94157d4d1fffa8c77fab501d6839de8be1b","revisions":{"92fdf94157d4d1fffa8c77fab501d6839de8be1b":{"kind":"REWORK","_number":1,"created":"2020-09-28 15:48:41.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/74/754774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/74/754774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/74/754774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/74/754774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/74/754774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/74/754774/1"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 15:48:40.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 15:48:40.000000000","tz":0},"subject":"Update master for stable/victoria","message":"Update master for stable/victoria\n\nAdd file to the reno documentation build to show release notes for\nstable/victoria.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/victoria.\n\nChange-Id: I0c20ebb46beb27640afc5258d7cabdfbda25c387\nSem-Ver: feature\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/92fdf94157d4d1fffa8c77fab501d6839de8be1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/92fdf94157d4d1fffa8c77fab501d6839de8be1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
