)]}'
{"id":"openstack%2Fmonasca-events-api~754965","triplet_id":"openstack%2Fmonasca-events-api~master~I133e5b7f26167451c8ccce9492ba750a440bc837","project":"openstack/monasca-events-api","branch":"master","hashtags":[],"change_id":"I133e5b7f26167451c8ccce9492ba750a440bc837","subject":"Replace assertItemsEqual with assertCountEqual","status":"ABANDONED","created":"2020-09-29 10:29:26.000000000","updated":"2020-09-30 04:52:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7ee10d1527638cc6e9c395a335c39378b0d0aab","_number":754965,"virtual_id_number":754965,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-30 02:44:34.000000000","_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-29 10:49:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 10:49:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"316a222fecef0d1fe6c1d61dafe8e90bf30de358","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-29 10:29:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cc6a4a487f28974e3d5430ede6edce383f84c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 10:49:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da64d5e07a3744cf8527bcb6548049b3 : FAILURE in 3m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b7163afd974c1ab907e0ab9f14d950 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348d3b38920d44fa94caccfe1334ecd4 : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d0dde965857430f81b44d3fbbbc7448 : FAILURE in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df20df3a4efc41708220a6b240fe59e3 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b76fca54d1c469d93ee5424e3bb730d : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3816d8ff8bf34f01b68a35ee78d31d32 : SUCCESS in 3m 30s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/2c709e8e873445babdbec62914a74606 : FAILURE in 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"aaecf004d1ec013a865a35c62e57f6694b574b00","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-29 11:41:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb1e997a394e12e20d5b49d931bf1da50ea77ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 12:00:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e6ecd7983314df58a5e7d200eb423a4 : FAILURE in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ea1f0b84d1243ee938d42ceaf19a4e0 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33226d68311548cd889ec147072c6419 : FAILURE in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46f9e891ad8f48509a35d89be5b5e415 : FAILURE in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f025df0a6fd64d61a99d0dcbd4aa9246 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f96c2dec2004a3786192d486fe8b9c8 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f89cc28ba1b444ccafc02cfc1b6db1d2 : SUCCESS in 3m 12s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/3cb01b81dbef4cf8a5e2f903f3248a42 : FAILURE in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"013c867a5f7a5573b1cfc780ce21137d3add1e3e","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-30 02:44:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dc940a413830e6ca1525d4465f3d0d01c92b162f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 03:06:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a32391ebb2f94ab29c0d8042ec66f681 : FAILURE in 2m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b415ea37d6f54c41806bb1f6234ef3e4 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bec213f8955441f8b2d930b14898287 : FAILURE in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34dea98060df45a0a51df32b3411a66b : FAILURE in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7c34c019f894cae8901d8632061e104 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e67ef653175a43efa690740b303ed1ab : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e4a7fb7bf14e239f601273fe09dd91 : SUCCESS in 3m 25s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/da70f76d69ae45ea81177c457826f78e : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a7ee10d1527638cc6e9c395a335c39378b0d0aab","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-30 04:52:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27f7a3f329429910e9e736a0f15d37a1b06d49f3","revisions":{"27f7a3f329429910e9e736a0f15d37a1b06d49f3":{"kind":"REWORK","_number":1,"created":"2020-09-29 10:29:26.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/65/754965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/65/754965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/754965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/754965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/754965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/65/754965/1"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-29 10:28:47.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-29 10:29:22.000000000","tz":480},"subject":"Replace assertItemsEqual with assertCountEqual","message":"Replace assertItemsEqual with assertCountEqual\n\nassertItemsEqual was removed from Python\u0027s unittest.TestCase in\nPython 3.3 [1][2]. We have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: I133e5b7f26167451c8ccce9492ba750a440bc837\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/27f7a3f329429910e9e736a0f15d37a1b06d49f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/27f7a3f329429910e9e736a0f15d37a1b06d49f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
