)]}'
{"id":"openstack%2Fmonasca-events-api~757425","triplet_id":"openstack%2Fmonasca-events-api~master~Idb14c8ed4f697c0e633c16e951f7c2ce88b86a3e","project":"openstack/monasca-events-api","branch":"master","hashtags":[],"change_id":"Idb14c8ed4f697c0e633c16e951f7c2ce88b86a3e","subject":"bump py37 to py38 in tox.ini","status":"ABANDONED","created":"2020-10-12 05:41:08.000000000","updated":"2026-02-03 22:32:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4bdbf81ba9206d5b18234dd95fcbcd9e658e7f91","_number":757425,"virtual_id_number":757425,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-03 00:53:54.000000000","_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-12 06:07:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 06:07:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 22:32:44.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"1beb5d84d23f6a9baf6700793d5c4c26c5ccc139","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-10-12 05:41:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0d51a0e462697482002286b6787c7db507ed1f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 06:07:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e2dd128f8acc43ff83df8af0065945f4 : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/670be2a4642248b1adfa7e4b089af8d4 : FAILURE in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17558971b7f046e9975c3e487ec00006 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b445f02339984143a3591fbdf9198511 : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec5f7403a94a411fa58ad935d3fc9698 : FAILURE in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b0747ac0d464a0a9932fdc865adfb72 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14c67ea2eb694ce0a6be8441bb330102 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ed476458044cfcbee4c9474717d93d : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d1c5315a4fb47cab9dcd277f790486f : SUCCESS in 3m 43s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/98a5c859f75f4ec9983f2a2fc6af6958 : FAILURE in 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"74785cea207bc1eb137002a0d3dada8261ed9d68","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-11-03 00:53:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3e3b558b84afe3d66bf81bd3d92b0db9655d418e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 01:13:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a864b9d40eb34837b88cfef15cd987e3 : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93c5fb2f906b42108265c930e593ab49 : FAILURE in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e06dcd2f869d45f8aff34861272cad0f : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4601823235748ff8a96024e6af61d8d : FAILURE in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b41ec2f07c54840b1da07f7183d6794 : FAILURE in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8338fad720284afebaaa065737cad7a0 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd8202f4e40b4e44a2d49a9c4f555b39 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2503396f3692416f97cdd86c484307cf : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e75531e06df34d0797fb54549aae4a38 : SUCCESS in 3m 19s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/9c7fdcb482e14e52b4eaedc86601d66f : FAILURE in 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4bdbf81ba9206d5b18234dd95fcbcd9e658e7f91","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:32:44.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"009baaaf152066a9cfb4c108086c7c80bb84a789","revisions":{"009baaaf152066a9cfb4c108086c7c80bb84a789":{"kind":"REWORK","_number":1,"created":"2020-10-12 05:41:08.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/25/757425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/25/757425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/25/757425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/25/757425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/25/757425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/25/757425/1"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-10-12 05:40:32.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-10-12 05:40:56.000000000","tz":480},"subject":"bump py37 to py38 in tox.ini","message":"bump py37 to py38 in tox.ini\n\nin \u0027victoria\u0027 cycle, we should test py38 by default.\n\nref:\n  https://governance.openstack.org/tc/reference/runtimes/victoria.html\n\nChange-Id: Idb14c8ed4f697c0e633c16e951f7c2ce88b86a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/009baaaf152066a9cfb4c108086c7c80bb84a789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/009baaaf152066a9cfb4c108086c7c80bb84a789"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
