)]}'
{"id":"openstack%2Fmonasca-events-api~787072","triplet_id":"openstack%2Fmonasca-events-api~master~I3ba7351e2bf74917277d0b627a0b2318bbb1501b","project":"openstack/monasca-events-api","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2026-02-03 22:31:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3ba7351e2bf74917277d0b627a0b2318bbb1501b","subject":"Use py3 as the default runtime for tox","status":"ABANDONED","created":"2021-04-20 05:43:49.000000000","updated":"2026-02-03 22:31:51.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a914bfbc92906eea14b2b75f177748a409103e2a","_number":787072,"virtual_id_number":787072,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-20 06:25:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2023-08-31 18:47:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 06:25:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 18:47:02.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-02-03 22:31:51.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"4f7e7c238bc81ff035baffcaf875eb4352a60d5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-04-20 05:43:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9810f3bd5296dfbe4d579d652e707332d193b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 06:25:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d754b40ca8e34cfd95ee8c3ef5901372 : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b760fc2137214173813b6b272b63922a : FAILURE in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/787f61713c154b23b6eb8ec6394ce435 : FAILURE in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b500a21e7f84599a195428bf952e1ef : FAILURE in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdc66dd72db0468aa25e678d1d75ede1 : FAILURE in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e616cab434447dc9a7d4e8919ef1a42 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42f8ebc6ceca47f6b9a395259ea6d6bd : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba8245af57c4502be2f69dfbf035cf5 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79bdd66c54b5472cb2c2e2ff4ec7ee38 : SUCCESS in 3m 50s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/31e79100464640f1915f2cd058f3d1de : FAILURE in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"18f41c105809a147605b64d7b6cf5776f0eb215e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-31 18:47:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20a90855bb74e3b99783d241cd1a5b5d04022bf7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-31 18:47:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf3564408cd10f9d14559b5b89dc2fb7a3bb9f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:30:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/790065bb9efd423baae4ab4d61087569\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/afbbffdb58df444d8f0fc52c4a7ffa14 : FAILURE in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cb05a01374043a684efc783630eb164 : FAILURE in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96d0f7aafac942e19e1b3ef2bd4ce8c4 : FAILURE in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c88f2cd5aed4f60aa23f36a86fd6700 : FAILURE in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4a7f5ae97cf4ce297e7b74cc3172294 : FAILURE in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e5cc98396534b90977dcbd80d06f892 : FAILURE in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/784c1a9c3393434588baadea3134c9cf : FAILURE in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aadb0276caa423ab3ff9b2a5d23efec : FAILURE in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22a8dbb2ad8f469ab7bf6826cbbfcd63 : SUCCESS in 5m 04s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/7c402dd5d31b4c22b7e603ab9e9f9144 : FAILURE in 37m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a914bfbc92906eea14b2b75f177748a409103e2a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:31:51.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6246db197545c07244b6a4d540ce4b83dbf3c36d","revisions":{"6246db197545c07244b6a4d540ce4b83dbf3c36d":{"kind":"REWORK","_number":1,"created":"2021-04-20 05:43:49.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/72/787072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/72/787072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/72/787072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/72/787072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/72/787072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/72/787072/1"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-04-20 05:43:20.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-04-20 05:43:47.000000000","tz":0},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update this at\neach new cycle.\n\nWallaby support officially the following runtimes [1]:\n- Python 3.6\n- Python 3.8\n\nDuring Victoria Python 3.7 was used as the default runtime [2] however this\nversion isn\u0027t longer officially supported.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html#python-runtimes-for-wallaby\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: I3ba7351e2bf74917277d0b627a0b2318bbb1501b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6246db197545c07244b6a4d540ce4b83dbf3c36d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/6246db197545c07244b6a4d540ce4b83dbf3c36d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
