)]}'
{"id":"openstack%2Fmonasca-events-api~859641","triplet_id":"openstack%2Fmonasca-events-api~master~I01f2b5033aed9d97758f32a00c61e5fc8f73f7fa","project":"openstack/monasca-events-api","branch":"master","topic":"fix-zuul-errors","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-06-15 06:53:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I01f2b5033aed9d97758f32a00c61e5fc8f73f7fa","subject":"Fix deprecated zuul queue syntax","status":"MERGED","created":"2022-09-28 15:48:15.000000000","updated":"2023-06-15 06:58:22.000000000","submitted":"2023-06-15 06:53:44.000000000","submitter":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859641-fix-zuul-errors","meta_rev_id":"6cc411fa122272949f913e3ef50be7dabaa1b66e","_number":859641,"virtual_id_number":859641,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-15 06:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-28 16:26:42.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-15 06:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-15 06:53:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-28 16:26:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-15 06:53:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"67c1a6c05e29c0a2dcc5d4475fd503538a651fdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-28 15:48:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b97dc8c309ff0359905c37dbb01f52e5047a3ef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 16:26:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bf59d4603c843fa923896a201a3debb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3452cd072b114ae89068f9af8cb7126d : FAILURE in 3m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/752be59d6eb9489983412d6a25a6e059 : FAILURE in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223ba9097a6b456cb9adb0a9804a392f : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89dbb260d22a4391be45312714736fc3 : FAILURE in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c89ec3696cc4b80bd7f92f736b5950e : FAILURE in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c0b8b936e994240a49aed4d77d24baa : SUCCESS in 3m 39s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/940ccfb69d8144fcb0ca4a417395f1e4 : FAILURE in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fb88c301116433e8958cf16f29660d1 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/526f5575d95d4126a169812553762d41 : SUCCESS in 3m 23s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/d84caf3c16c848dfbb754f84082c782a : FAILURE in 36m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b9a5940cb048148902781d2d388f3c8a66c7f1f9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 06:48:46.000000000","message":"Topic changed from fix-zuul-queue-syntax to fix-zuul-errors","accounts_in_message":[],"_revision_number":1},{"id":"4d51b23b9a63086c664fe231fd3402d20e5ceb45","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 06:53:33.000000000","message":"Patch Set 1: Verified+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0656f2d3c70f4ddb9539b6eb558d7ed1bc558f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 06:53:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6cc411fa122272949f913e3ef50be7dabaa1b66e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 06:58:22.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3000a0e67b884d8aa4221b50b011e5ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0c572df49a44fcb0e1123f2f44a18d : FAILURE in 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04633984a03e901d077b794b7c185a87756595a7","revisions":{"04633984a03e901d077b794b7c185a87756595a7":{"kind":"REWORK","_number":1,"created":"2022-09-28 15:48:15.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/859641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/41/859641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/41/859641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/41/859641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/41/859641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/41/859641/1"}}},"commit":{"parents":[{"commit":"fc97ceed7af98b196cfc157cb1b7a69101d2970a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/fc97ceed7af98b196cfc157cb1b7a69101d2970a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-28 15:47:03.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-28 15:48:08.000000000","tz":120},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work for\nmonasca-events-api.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n\nChange-Id: I01f2b5033aed9d97758f32a00c61e5fc8f73f7fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/04633984a03e901d077b794b7c185a87756595a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/04633984a03e901d077b794b7c185a87756595a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
