)]}'
{"id":"openstack%2Fmonasca-events-api~922268","triplet_id":"openstack%2Fmonasca-events-api~master~Iac2c19714da5f2b8abadca2e7c84f66f0f0d1654","project":"openstack/monasca-events-api","branch":"master","topic":"retire-devstack-gate","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2024-06-20 11:10:18.000000000","reason":"Change was submitted"},"37102":{"account":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"last_update":"2024-06-20 11:10:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac2c19714da5f2b8abadca2e7c84f66f0f0d1654","subject":"Drop devstack-gate","status":"MERGED","created":"2024-06-19 03:03:41.000000000","updated":"2024-06-20 11:13:35.000000000","submitted":"2024-06-20 11:10:18.000000000","submitter":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922268-retire-devstack-gate","meta_rev_id":"d1899d9c8bdf925558df6f9496ce7e3ac655c574","_number":922268,"virtual_id_number":922268,"owner":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-19 03:46:36.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-20 11:10:18.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-06-20 11:07:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-20 11:10:18.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-20 11:10:18.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-19 03:46:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-19 03:48:54.000000000","updated_by":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"},{"updated":"2024-06-20 11:07:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-06-20 11:10:18.000000000","updated_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"reviewer":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"state":"REVIEWER"}],"messages":[{"id":"e142705b9afd439914a6f57ac57ae1f11b5fd8d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"date":"2024-06-19 03:03:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd78a86af1ebb3273eb8247fdb9e66f47a326be8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 03:46:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a76215c72c24d62879f7373a8427915\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fcef36b979e41eba975bd98bf090000 : FAILURE in 2m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/511b16c1f6584fcaaad6651935e7f472 : FAILURE in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b42bcd364624e4798b0d4ac35f4539d : FAILURE in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e27734d0acec4fc88e570302566ef81c : FAILURE in 2m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c57ef3641fad487fa41a07eddd94e20e : FAILURE in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7392b559e764465ba2795d09ae6fece8 : FAILURE in 1m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29a70948b6d48d5bfe3416b4de9acf5 : FAILURE in 2m 48s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/73b854e18ed340a3a4fc27ff25753784 : FAILURE in 37m 26s","accounts_in_message":[],"_revision_number":1},{"id":"505e2ef8480359d6c4231e4fd44e4055bdc19095","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-20 11:07:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df31c7b2adef43baecb6c10e13a89c254d2b6e41","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2024-06-20 11:10:18.000000000","message":"Patch Set 1: Code-Review+2 Verified+2 Workflow+1\n\nBypassing Zuul to merge this and unblock devstack-gate cleanup.","accounts_in_message":[],"_revision_number":1},{"id":"6ab3e5af6b2ae65d884651741ba49cb9e24f7a4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2024-06-20 11:10:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d1899d9c8bdf925558df6f9496ce7e3ac655c574","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 11:13:35.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3488d3cc0eb42b5abe44d4bcf5f4dc6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3cd3b434874c53a568f48ee6d92533 : FAILURE in 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","revisions":{"f463084bf08a9c5a23b533bffb36af5eec3ae07f":{"kind":"REWORK","_number":1,"created":"2024-06-19 03:03:41.000000000","uploader":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"ref":"refs/changes/68/922268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/68/922268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/68/922268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/68/922268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/68/922268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/68/922268/1"}}},"commit":{"parents":[{"commit":"04633984a03e901d077b794b7c185a87756595a7","subject":"Fix deprecated zuul queue syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/04633984a03e901d077b794b7c185a87756595a7"}]}],"author":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-06-19 03:02:39.000000000","tz":-240},"committer":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-06-19 03:03:40.000000000","tz":-240},"subject":"Drop devstack-gate","message":"Drop devstack-gate\n\nThe devstack-gate repo has been retired, let\u0027s drop all references\nto it to avoid errors like this in Zuul:\n\nThe project \"openstack/devstack-gate\" was not found.\n\nChange-Id: Iac2c19714da5f2b8abadca2e7c84f66f0f0d1654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
