)]}'
{"id":"openstack%2Fmonasca-events-api~942917","triplet_id":"openstack%2Fmonasca-events-api~master~Ic18808adf70bac8dbc393b01c5a0019737368db0","project":"openstack/monasca-events-api","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-02-02 15:02:32.000000000","reason":"Change was abandoned"},"35988":{"account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"last_update":"2025-03-01 23:03:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic18808adf70bac8dbc393b01c5a0019737368db0","subject":"Update tox commands and address bit rot","status":"ABANDONED","created":"2025-02-27 15:56:14.000000000","updated":"2026-02-02 15:02:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b82ffd39ac6ef7c0344348b0222d1ce1224e5c53","_number":942917,"virtual_id_number":942917,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-01 23:00:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"all":[{"value":2,"date":"2025-03-01 23:03:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-27 19:34:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"state":"REVIEWER"},{"updated":"2025-02-27 22:17:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9bbefada443f946d6d082de38c3d72b609dc4dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-27 15:56:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28d372fe95ca0e532c31b6aca4b74f9e00b7d124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-27 17:26:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e81d6bcde9705b9edcb4f54e11877655366243e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 22:17:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9121eb809a8f451eb3e123f42429962a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6d07a62f2d234f78b80f65dcd877ba9a : SUCCESS in 2m 58s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bbc44691c5604ab889b60e702d6a7ee3 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8945cc0a4ce48c5a5756ef6a9e9e9e8 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d53bc25c13f47c5b82b30e7fab2f93d : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d5af07d5a42452cb9f21d3b48a66a63 : SUCCESS in 5m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/33c38b8b9ccf44c5ab96b9af26bc83c7 : FAILURE in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6102dfb9444e499cb7da9b7c63233b28 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d77e9e9fba9946ab89a410c1278b7850 : SUCCESS in 3m 07s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/71e02eac4d11410e92751c2ed767b54e : FAILURE in 21m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e520bd0bc365faff1353e1f24425ae0a7c9b4241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-28 15:28:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8d21336225bc0651469943f34c294ecd307bc8d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-28 15:29:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9dc1a3dabda9a305f4c4141020eaf72ef5a3488e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 15:29:56.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/monasca-events-api on branch stable/2023.2.  The\nproblem was:\n\n  Job monasca-events-tempest-base not defined\n\nThe problem appears in the the \"monasca-events-python3\" job stanza:\n\n  job:\n      name: monasca-events-python3\n      parent: monasca-events-tempest-base\n      vars:\n        devstack_localrc:\n          USE_PYTHON3: true\n          TEMPEST_PLUGINS: /opt/stack/monasca-tempest-plugin\n\n  in \"openstack/monasca-events-api/.zuul.yaml@stable/2023.2\", line 38","accounts_in_message":[],"_revision_number":4},{"id":"e55f5825205830b91fa80b9672281ed3187fdcdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-28 16:54:05.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fc9c14f1d9dcd95fcbddf2d49ca10e6d73f4e639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 16:54:36.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/monasca-events-api on branch stable/2023.2.  The\nproblem was:\n\n  Job monasca-events-tempest-base not defined\n\nThe problem appears in the the \"monasca-events-python3\" job stanza:\n\n  job:\n      name: monasca-events-python3\n      parent: monasca-events-tempest-base\n      vars:\n        devstack_localrc:\n          USE_PYTHON3: true\n          TEMPEST_PLUGINS: /opt/stack/monasca-tempest-plugin\n\n  in \"openstack/monasca-events-api/.zuul.yaml@stable/2023.2\", line 38","accounts_in_message":[],"_revision_number":5},{"id":"ddb2b063e579c7299a472ed812291aaf52699b42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-28 16:56:54.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"dac1b9891566c4619afe1d2b9ce92529e0569ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 16:57:27.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/monasca-events-api on branch stable/2023.2.  The\nproblem was:\n\n  Job monasca-events-tempest-base not defined\n\nThe problem appears in the the \"monasca-events-python3\" job stanza:\n\n  job:\n      name: monasca-events-python3\n      parent: monasca-events-tempest-base\n      vars:\n        devstack_localrc:\n          USE_PYTHON3: true\n          TEMPEST_PLUGINS: /opt/stack/monasca-tempest-plugin\n\n  in \"openstack/monasca-events-api/.zuul.yaml@stable/2023.2\", line 38","accounts_in_message":[],"_revision_number":6},{"id":"2ad8e1317d6dea0290563e7ee075db8152b641ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-28 16:59:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"80d400c1945911828cbc497179ef241ac0212a10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-02-28 17:32:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d362542093b180eb2adf720c07463da9547be390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 18:09:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4805216488a64c23b22e3755c56f0ae8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3e4caeb6af63457ea9c27414d64ffd4e : FAILURE in 6m 54s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1959fa6281c7495293e180186e26405d : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba4b4d7606cf4cc78c510826b663e9b4 : FAILURE in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2cdb429f2a0400690fa7bbc13b8af49 : SUCCESS in 7m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79cf1278dfe548b6b9e93b7baaf3a96a : SUCCESS in 4m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0e5d3d16f024731a186a0ad1cc2168d : SUCCESS in 1m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a924275b5be543308d6f4525e6f54585 : FAILURE in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e306a7abebbf4bfd8b8420776e306be7 : SUCCESS in 3m 31s\n- monasca-events-api-py3 https://zuul.opendev.org/t/openstack/build/4489c79c70b1418ea028795b2c7fe01b : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":8},{"id":"885d9aed1df76ad0ac5a972be9e4082bb3c842a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2025-03-01 22:49:22.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"68f1e4e714c992d2407264e7774432a45751863a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 23:00:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818ccfa66540411ea5cad01a66ff6786\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c69b722caf134fb4992b9d5060176eea : SUCCESS in 1m 56s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c0d6ab56745496ea0f9eba3d1cbb83a : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f1b65c9abb44bd8253567884e41885 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50a8aad10fc84a709583fe64a355b869 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc0fe9f68dac47c997cff667dc333371 : SUCCESS in 2m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/249f45a7945f48cf844a3771e2f58692 : SUCCESS in 1m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee1f521c77248f7a69df8d4855b3aa5 : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbd50550158c471ea718b8c89c97459b : SUCCESS in 2m 07s\n- monasca-events-api-py3 https://zuul.opendev.org/t/openstack/build/e956ad8864f04c0ead24b45aedeed02c : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":9},{"id":"d5efb3b0a000e77a79f8f1771e7e82d84ccf865a","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2025-03-01 23:03:26.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"b82ffd39ac6ef7c0344348b0222d1ce1224e5c53","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-02 15:02:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e1a7a78b6477837d3c01f8a7d243cca5954355dc","revisions":{"88da47e34191117e6ca18a88a6c112789510f225":{"kind":"REWORK","_number":1,"created":"2025-02-27 15:56:14.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/942917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/1"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:56:09.000000000","tz":-360},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/88da47e34191117e6ca18a88a6c112789510f225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/88da47e34191117e6ca18a88a6c112789510f225"}]},"branch":"refs/heads/master"},"85a584f856b093ec0e1350df5313cf5be8fcd36f":{"kind":"REWORK","_number":2,"created":"2025-02-27 17:26:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/942917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/2"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 17:25:54.000000000","tz":-360},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/85a584f856b093ec0e1350df5313cf5be8fcd36f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/85a584f856b093ec0e1350df5313cf5be8fcd36f"}]},"branch":"refs/heads/master"},"dc3118924f9179227b948798d663a61823b8d91f":{"kind":"REWORK","_number":3,"created":"2025-02-28 15:28:30.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/942917/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/3"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-28 15:26:49.000000000","tz":-360},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nDrops devstack test. Devstack gate testing was dropped in 0463398 and\nthis builds on that.\n\nAlso dropping lower constraints to follow what many other projects have\ndone. It is a useful test to validate the minimum version requirements,\nbut unless and until more maintenance can be done to handle things\nbetter, we are better off just going with the default upper constraints.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/dc3118924f9179227b948798d663a61823b8d91f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/dc3118924f9179227b948798d663a61823b8d91f"}]},"branch":"refs/heads/master"},"a11970fb1795cda4c092c76abce5b49cfddba3f8":{"kind":"REWORK","_number":4,"created":"2025-02-28 15:29:16.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/942917/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/4"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-28 15:29:11.000000000","tz":-360},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nDrops devstack test. Devstack gate testing was dropped in 0463398 and\nthis builds on that.\n\nAlso dropping lower constraints to follow what many other projects have\ndone. It is a useful test to validate the minimum version requirements,\nbut unless and until more maintenance can be done to handle things\nbetter, we are better off just going with the default upper constraints.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/a11970fb1795cda4c092c76abce5b49cfddba3f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/a11970fb1795cda4c092c76abce5b49cfddba3f8"}]},"branch":"refs/heads/master"},"5f15eaabe619bfd5d6d07fd237b856d88eb0c93d":{"kind":"REWORK","_number":5,"created":"2025-02-28 16:54:05.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/942917/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/5"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-28 16:53:58.000000000","tz":-360},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nDrops devstack test. Devstack gate testing was dropped in 0463398 and\nthis builds on that.\n\nAlso dropping lower constraints to follow what many other projects have\ndone. It is a useful test to validate the minimum version requirements,\nbut unless and until more maintenance can be done to handle things\nbetter, we are better off just going with the default upper constraints.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/5f15eaabe619bfd5d6d07fd237b856d88eb0c93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/5f15eaabe619bfd5d6d07fd237b856d88eb0c93d"}]},"branch":"refs/heads/master"},"51ea592937cc32d4bd5c66bce786f752658dd3c7":{"kind":"REWORK","_number":6,"created":"2025-02-28 16:56:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/942917/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/6"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-28 16:56:48.000000000","tz":-360},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nDrops devstack test. Devstack gate testing was dropped in 0463398 and\nthis builds on that.\n\nAlso dropping lower constraints to follow what many other projects have\ndone. It is a useful test to validate the minimum version requirements,\nbut unless and until more maintenance can be done to handle things\nbetter, we are better off just going with the default upper constraints.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/51ea592937cc32d4bd5c66bce786f752658dd3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/51ea592937cc32d4bd5c66bce786f752658dd3c7"}]},"branch":"refs/heads/master"},"621fe0d16427a08cb5b5e26a7495b79b2f328eed":{"kind":"REWORK","_number":7,"created":"2025-02-28 16:59:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/942917/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/7"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-28 16:59:54.000000000","tz":-360},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nDrops devstack test. Devstack gate testing was dropped in 0463398 and\nthis builds on that.\n\nAlso dropping lower constraints to follow what many other projects have\ndone. It is a useful test to validate the minimum version requirements,\nbut unless and until more maintenance can be done to handle things\nbetter, we are better off just going with the default upper constraints.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/621fe0d16427a08cb5b5e26a7495b79b2f328eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/621fe0d16427a08cb5b5e26a7495b79b2f328eed"}]},"branch":"refs/heads/master"},"5d1b5266071e3078bddea1fddceb162098d4bd6e":{"kind":"REWORK","_number":8,"created":"2025-02-28 17:32:02.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/942917/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/8"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-28 17:31:54.000000000","tz":-360},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nDrops devstack test. Devstack gate testing was dropped in 0463398 and\nthis builds on that.\n\nAlso dropping lower constraints to follow what many other projects have\ndone. It is a useful test to validate the minimum version requirements,\nbut unless and until more maintenance can be done to handle things\nbetter, we are better off just going with the default upper constraints.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/5d1b5266071e3078bddea1fddceb162098d4bd6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/5d1b5266071e3078bddea1fddceb162098d4bd6e"}]},"branch":"refs/heads/master"},"e1a7a78b6477837d3c01f8a7d243cca5954355dc":{"kind":"REWORK","_number":9,"created":"2025-03-01 22:49:22.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/17/942917/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/17/942917/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/17/942917/9"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2025-02-27 15:53:45.000000000","tz":-360},"committer":{"name":"Hasan ACAR","email":"hasan.acar@tubitak.gov.tr","date":"2025-03-01 22:49:21.000000000","tz":0},"subject":"Update tox commands and address bit rot","message":"Update tox commands and address bit rot\n\nThis gets things updated to start passing CI again. It cleans up some\nimports and some old py2 handling that is not needed anymore, and\nupdates the tox configuration to use the current recommendations.\n\nOne unit test is failing due to changes in oslo_policy support. This was\na bit too much for me to take on with a drive by patch, but does need to\nbe addressed by someone more familiar with the code in the (hopefully\nnear) future.\n\nDrops devstack test. Devstack gate testing was dropped in 0463398 and\nthis builds on that.\n\nAlso dropping lower constraints to follow what many other projects have\ndone. It is a useful test to validate the minimum version requirements,\nbut unless and until more maintenance can be done to handle things\nbetter, we are better off just going with the default upper constraints.\n\nChange-Id: Ic18808adf70bac8dbc393b01c5a0019737368db0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/e1a7a78b6477837d3c01f8a7d243cca5954355dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/e1a7a78b6477837d3c01f8a7d243cca5954355dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
