)]}'
{"id":"openstack%2Fmonasca-events-api~953505","triplet_id":"openstack%2Fmonasca-events-api~master~I1e77dd74aaea385b8e37066b61d414fb6d4e31bf","project":"openstack/monasca-events-api","branch":"master","topic":"remove-USE_PYTHON3","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-02-03 22:31:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1e77dd74aaea385b8e37066b61d414fb6d4e31bf","subject":"Zuul: do not use USE_PYTHON3","status":"ABANDONED","created":"2025-06-27 00:42:55.000000000","updated":"2026-02-03 22:31:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8ebf02595c7bb28184c29b068b54d72b6042b9a3","_number":953505,"virtual_id_number":953505,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-27 01:03:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 01:03:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 22:31:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"e62e700785c5037288d5ad8adcb7416b1e4d0fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-27 00:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71d1f6d3ad8330adb879cd2c280d2a28c340008f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 01:03:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b903acec4b141f398e58e2f9c7d24bb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f40967d8c1bb4d028033266b419bd931 : FAILURE in 2m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79cbbcdd8d724c7babaf591908eb58da : FAILURE in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f606d85489de41c8a92755040d2b2c51 : FAILURE in 2m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/342cef4f04e34862ad09d017cc4a3103 : FAILURE in 2m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8634486addb446b2b6b3b05e59b840f8 : FAILURE in 1m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d7b4b194db74382b174f5bc785c6ee7 : FAILURE in 2m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f611350497954375b40a0f8ab45504c8 : FAILURE in 2m 19s\n- monasca-events-python3 https://zuul.opendev.org/t/openstack/build/22449d33760147df9e339b3ff10e615d : FAILURE in 18m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8ebf02595c7bb28184c29b068b54d72b6042b9a3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:31:34.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f00162c61935bd3d4a03083c21983083983c44a","revisions":{"7f00162c61935bd3d4a03083c21983083983c44a":{"kind":"REWORK","_number":1,"created":"2025-06-27 00:42:55.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/05/953505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/05/953505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/05/953505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/05/953505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/05/953505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/05/953505/1"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-27 00:42:39.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-27 00:42:53.000000000","tz":120},"subject":"Zuul: do not use USE_PYTHON3","message":"Zuul: do not use USE_PYTHON3\n\nDevstack has removed the USE_PYTHON3 variable[1][2] and now always uses\nPython 3.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/920658\n[2] Commit 5412dbfe7b797149f1f68100de8003b1876398fe\n\nChange-Id: I1e77dd74aaea385b8e37066b61d414fb6d4e31bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/7f00162c61935bd3d4a03083c21983083983c44a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/7f00162c61935bd3d4a03083c21983083983c44a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
