)]}'
{"id":"openstack%2Fmonasca-events-api~957065","triplet_id":"openstack%2Fmonasca-events-api~master~I2687afaaefb9ab13d81f2f2cb536495945d47e7b","project":"openstack/monasca-events-api","branch":"master","topic":"retire-monasca-events-api","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-11-24 17:05:32.000000000","reason":"Change was submitted"}},"hashtags":["monasca-retirement"],"change_id":"I2687afaaefb9ab13d81f2f2cb536495945d47e7b","subject":"Retire monasca-events-api repository","status":"MERGED","created":"2025-08-12 05:18:11.000000000","updated":"2025-11-24 17:05:32.000000000","submitted":"2025-11-24 17:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957065-retire-monasca-events-api","meta_rev_id":"4f5d5fc223985b77a02d1433b98036c5a722d9fd","_number":957065,"virtual_id_number":957065,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-24 17:05:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-11-24 17:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"date":"2025-10-20 21:56:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-11-24 17:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 05:19:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 15:35:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-11-24 17:04:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"dfd7a3edf40ead04efcf3866eb3cee067f9ca626","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-12 05:18:11.000000000","message":"Hashtag added: monasca-retirement","accounts_in_message":[],"_revision_number":1},{"id":"546911c38162a9192d90c004d5bdaa1db88d9dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 05:19:51.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/monasca-events-api on branch unmaintained/2023.1.  The\nproblem was:\n\n  Job monasca-events-tempest-base not defined\n\nThe problem appears in the the \"monasca-events-python3\" job stanza:\n\n  job:\n      name: monasca-events-python3\n      parent: monasca-events-tempest-base\n      vars:\n        devstack_localrc:\n          USE_PYTHON3: true\n          TEMPEST_PLUGINS: /opt/stack/monasca-tempest-plugin\n\n  in \"openstack/monasca-events-api/.zuul.yaml@unmaintained/2023.1\", line 38","accounts_in_message":[],"_revision_number":1},{"id":"e35ba88cbb7ad94192cd374810dc19a0e0d4a3d0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-13 21:33:28.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"87a8ba2510b1c83afe76e76bb5526c298260204e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-15 18:28:35.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c6d3f99fba14083af88e68f1f19dff27679be166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 18:28:48.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/monasca-events-api on branch unmaintained/2023.1.  The\nproblem was:\n\n  Job monasca-events-tempest-base not defined\n\nThe problem appears in the the \"monasca-events-python3\" job stanza:\n\n  job:\n      name: monasca-events-python3\n      parent: monasca-events-tempest-base\n      vars:\n        devstack_localrc:\n          USE_PYTHON3: true\n          TEMPEST_PLUGINS: /opt/stack/monasca-tempest-plugin\n\n  in \"openstack/monasca-events-api/.zuul.yaml@unmaintained/2023.1\", line 38","accounts_in_message":[],"_revision_number":2},{"id":"ef237c3532db2779aa99fd6fd7bb61fc7dd9c875","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-20 17:02:28.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0196be10f51d005cd9fc01bd418d84b82ff26c28","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-20 21:00:00.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"3a6cfd3fc480f9e5c10ee034910c7a4d7cdf9871","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-10-20 21:56:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56b15eb8ce02d963e70ac8256ecbac3397af2dbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 21:57:28.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/monasca-events-api on branch unmaintained/2023.1.  The\nproblem was:\n\n  Job monasca-events-tempest-base not defined\n\nThe problem appears in the the \"monasca-events-python3\" job stanza:\n\n  job:\n      name: monasca-events-python3\n      parent: monasca-events-tempest-base\n      vars:\n        devstack_localrc:\n          USE_PYTHON3: true\n          TEMPEST_PLUGINS: /opt/stack/monasca-tempest-plugin\n\n  in \"openstack/monasca-events-api/.zuul.yaml@unmaintained/2023.1\", line 38","accounts_in_message":[],"_revision_number":2},{"id":"4d5d9532fb92dc9cc93d0af09187ab9c6abefe4a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-24 15:35:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e77366e1ffbdff0c57457b68a6349fb030c12e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 15:37:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f8f4ef843846ceb21d99898c4d76e4\n\n- noop https://zuul.opendev.org/t/openstack/build/85eb2918af2d4595a929b2f3c521fb8e : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"5516f0a643385eb9b14a6a8317fe5382653cd47a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-24 17:04:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da4c72c32369f8fa923894eead40ddf475e97975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:05:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2448a7c057f78d285b0865c40e263c0fc9b9c551","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:05:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76490bd0a6d64f3eab2d33cfe634e085\n\n- noop https://zuul.opendev.org/t/openstack/build/39b85925fd56423aa73bbf1a392b2d71 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4f5d5fc223985b77a02d1433b98036c5a722d9fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:05:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1179d7a55498101e78da8e7b3f0a4f4bf07eacd4","revisions":{"45d7a2f47ad9f39319d70194998bee720ab20977":{"kind":"REWORK","_number":1,"created":"2025-08-12 05:18:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/957065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/65/957065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/957065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/957065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/957065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/65/957065/1"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-12 05:15:05.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-12 05:18:09.000000000","tz":-420},"subject":"Retire monasca-events-api repository","message":"Retire monasca-events-api repository\n\nThis repository is being retired as part of the Monasca project\nretirement. The project content has been replaced with a retirement\nnotice.\n\nDepends-On: I3cb522ce8f51424b64e93c1efaf0dfd1781cd5ac\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/957063\n\nChange-Id: I2687afaaefb9ab13d81f2f2cb536495945d47e7b\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/45d7a2f47ad9f39319d70194998bee720ab20977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/45d7a2f47ad9f39319d70194998bee720ab20977"}]},"branch":"refs/heads/master"},"1179d7a55498101e78da8e7b3f0a4f4bf07eacd4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-15 18:28:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/957065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-events-api","ref":"refs/changes/65/957065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/957065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/957065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-events-api refs/changes/65/957065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-events-api refs/changes/65/957065/2"}}},"commit":{"parents":[{"commit":"f463084bf08a9c5a23b533bffb36af5eec3ae07f","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/f463084bf08a9c5a23b533bffb36af5eec3ae07f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-12 05:15:05.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-08-15 18:28:35.000000000","tz":0},"subject":"Retire monasca-events-api repository","message":"Retire monasca-events-api repository\n\nThis repository is being retired as part of the Monasca project\nretirement. The project content has been replaced with a retirement\nnotice.\n\nNeeded-By: I3cb522ce8f51424b64e93c1efaf0dfd1781cd5ac\nChange-Id: I2687afaaefb9ab13d81f2f2cb536495945d47e7b\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/1179d7a55498101e78da8e7b3f0a4f4bf07eacd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-events-api/commit/1179d7a55498101e78da8e7b3f0a4f4bf07eacd4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
