)]}'
{"id":"openstack%2Fmonasca-notification~519630","triplet_id":"openstack%2Fmonasca-notification~master~Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef","project":"openstack/monasca-notification","branch":"master","topic":"bp/templates-for-slack-notifications","hashtags":[],"change_id":"Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef","subject":"Templates for Slack notifications","status":"MERGED","created":"2017-11-14 12:09:55.000000000","updated":"2019-08-07 12:04:51.000000000","submitted":"2019-08-07 12:04:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"519630-1565179491074-0d575412","meta_rev_id":"f305fc7c2a168e626a93c5476bc89d57c6722ab6","_number":519630,"virtual_id_number":519630,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"date":"2019-05-22 09:41:08.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-07 12:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":2,"date":"2019-08-07 11:53:41.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-23 10:44:24.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":1,"date":"2019-08-07 11:53:41.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-14 12:12:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2017-11-14 12:18:42.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2019-07-23 10:44:24.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-08-07 11:53:41.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-08-07 12:04:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"870ac3a1a6e55bf1b6859bbb817423156ecd0222","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2017-11-14 12:09:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53088516938a830ac42c6a18c58e7bb994b0dafe","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2017-11-14 12:18:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nIt\u0027s a big improvement in usability.  A small nit on wording but only if you\u0027re respinning.","accounts_in_message":[],"_revision_number":1},{"id":"350638a5efa31ee41dd927ced300f5acc8e9093c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2017-11-14 12:31:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b7dbc493021ccf35d599f3ca245d1b0756bf2b0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2017-11-14 12:37:28.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for the review - one comment inline.","accounts_in_message":[],"_revision_number":1},{"id":"79573125a6855606a53e24f9d58bbdf97c288ec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-14 13:37:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/519630/2/check/openstack-tox-pep8/3120d9c/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/30/519630/2/check/openstack-tox-py27/b580f4e/ : SUCCESS in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/30/519630/2/check/openstack-tox-py35/81fb049/ : SUCCESS in 3m 49s\n- openstack-tox-cover http://logs.openstack.org/30/519630/2/check/openstack-tox-cover/a94d130/cover/ : SUCCESS in 3m 38s\n- monasca-tempest-python-mysql http://logs.openstack.org/30/519630/2/check/monasca-tempest-python-mysql/56cfea1/ : SUCCESS in 1h 03m 33s (non-voting)\n- monasca-tempest-python-postgresql http://logs.openstack.org/30/519630/2/check/monasca-tempest-python-postgresql/4e5f234/ : FAILURE in 39m 11s (non-voting)\n- monasca-tempest-java-mysql http://logs.openstack.org/30/519630/2/check/monasca-tempest-java-mysql/d0357f0/ : SUCCESS in 48m 39s (non-voting)\n- monasca-tempest-java-postgresql http://logs.openstack.org/30/519630/2/check/monasca-tempest-java-postgresql/e2925ca/ : FAILURE in 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fe81ec6ea6539b5897e3683cd689f78800a0207","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2017-11-22 15:45:03.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"cb5df053c62914345c4b3c78880f10cb252c8f3a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2017-11-22 15:45:57.000000000","message":"Patch Set 3: Workflow-1\n\nNeed to add default template.","accounts_in_message":[],"_revision_number":3},{"id":"39caf6c435625dd773c422aa405b2c3b47a1977f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-22 16:44:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/519630/3/check/openstack-tox-pep8/3ad0880/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/30/519630/3/check/openstack-tox-py27/6de57ca/ : SUCCESS in 6m 16s\n- openstack-tox-py35 http://logs.openstack.org/30/519630/3/check/openstack-tox-py35/abae1df/ : SUCCESS in 4m 05s\n- openstack-tox-cover http://logs.openstack.org/30/519630/3/check/openstack-tox-cover/5cc3cb3/cover/ : SUCCESS in 4m 25s\n- monasca-tempest-python-mysql http://logs.openstack.org/30/519630/3/check/monasca-tempest-python-mysql/64f6f72/ : SUCCESS in 47m 24s (non-voting)\n- monasca-tempest-python-postgresql http://logs.openstack.org/30/519630/3/check/monasca-tempest-python-postgresql/cdf705f/ : FAILURE in 29m 24s (non-voting)\n- monasca-tempest-java-mysql http://logs.openstack.org/30/519630/3/check/monasca-tempest-java-mysql/2d17c63/ : SUCCESS in 56m 00s (non-voting)\n- monasca-tempest-java-postgresql http://logs.openstack.org/30/519630/3/check/monasca-tempest-java-postgresql/aa12e92/ : FAILURE in 49m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6012dc57a1ae60bbeb83714437e12b5fd1836804","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-30 12:32:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"58c32c0ed8689170c6c13472e32e6b65e1ac1bf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 13:18:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/30/519630/4/check/openstack-tox-cover/c01ded2/cover/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/519630/4/check/openstack-tox-lower-constraints/6577170/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/30/519630/4/check/openstack-tox-pep8/a28205e/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/30/519630/4/check/openstack-tox-py27/4cd3470/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/30/519630/4/check/openstack-tox-py36/651a02c/ : SUCCESS in 4m 56s\n- openstack-tox-py37 http://logs.openstack.org/30/519630/4/check/openstack-tox-py37/c50c31b/ : SUCCESS in 5m 17s\n- build-monasca-docker-image http://logs.openstack.org/30/519630/4/check/build-monasca-docker-image/ade5618/ : SUCCESS in 6m 11s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/30/519630/4/check/monasca-tempest-python2-influxdb/243f297/ : SUCCESS in 37m 35s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/30/519630/4/check/monasca-tempest-python3-influxdb/3daae5d/ : SUCCESS in 40m 55s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/30/519630/4/check/monasca-tempest-java-influxdb/0df0d90/ : FAILURE in 45m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f688163234ee51c6c5bd6ae6eff4a48b823d1c8a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-30 13:41:39.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"560d57cc3886169a27856295150416986a81167f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-01 13:29:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9db003341365b884849cce94dfd53004845ff2a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 14:16:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/30/519630/5/check/test-release-openstack/e23e881/ : FAILURE in 3m 00s\n- openstack-tox-cover http://logs.openstack.org/30/519630/5/check/openstack-tox-cover/4f130bd/cover/ : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/519630/5/check/openstack-tox-lower-constraints/8599054/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/30/519630/5/check/openstack-tox-pep8/113b624/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/30/519630/5/check/openstack-tox-py27/4d9a560/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/30/519630/5/check/openstack-tox-py36/3dcb257/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/30/519630/5/check/openstack-tox-py37/6b0cc19/ : SUCCESS in 4m 27s\n- build-monasca-docker-image http://logs.openstack.org/30/519630/5/check/build-monasca-docker-image/2dca08f/ : SUCCESS in 5m 59s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/30/519630/5/check/monasca-tempest-python2-influxdb/3ff3c54/ : SUCCESS in 40m 58s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/30/519630/5/check/monasca-tempest-python3-influxdb/cbe38d9/ : SUCCESS in 44m 32s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/30/519630/5/check/monasca-tempest-java-influxdb/b894566/ : FAILURE in 42m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"364870f6a196660754a8e14b23f36002dd01216f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-01 14:42:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"56c9a1ee5e58af39bfad7175f1fce423a218d98c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 15:38:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/30/519630/6/check/test-release-openstack/055650c/ : SUCCESS in 3m 21s\n- openstack-tox-cover http://logs.openstack.org/30/519630/6/check/openstack-tox-cover/b4e55a7/cover/ : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/519630/6/check/openstack-tox-lower-constraints/35c4ef4/ : SUCCESS in 7m 30s\n- openstack-tox-pep8 http://logs.openstack.org/30/519630/6/check/openstack-tox-pep8/4174f26/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/30/519630/6/check/openstack-tox-py27/17d54bf/ : SUCCESS in 6m 15s\n- openstack-tox-py36 http://logs.openstack.org/30/519630/6/check/openstack-tox-py36/c43d26a/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/30/519630/6/check/openstack-tox-py37/e7be583/ : SUCCESS in 6m 08s\n- build-monasca-docker-image http://logs.openstack.org/30/519630/6/check/build-monasca-docker-image/315cf5e/ : SUCCESS in 5m 24s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/30/519630/6/check/monasca-tempest-python2-influxdb/a589cf4/ : FAILURE in 41m 41s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/30/519630/6/check/monasca-tempest-python3-influxdb/c3dbd72/ : FAILURE in 31m 09s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/30/519630/6/check/monasca-tempest-java-influxdb/b514123/ : FAILURE in 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d12098ecee806b4cd455fda3f080c2c65f38043b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-02 08:53:17.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"73482bae7af70322a6695a94492c8d6972aa17cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 09:46:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/30/519630/7/check/test-release-openstack/7bb9446/ : SUCCESS in 2m 50s\n- openstack-tox-cover http://logs.openstack.org/30/519630/7/check/openstack-tox-cover/782770f/cover/ : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/519630/7/check/openstack-tox-lower-constraints/fe4dbf2/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/30/519630/7/check/openstack-tox-pep8/36c02b7/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/30/519630/7/check/openstack-tox-py27/9b9aa27/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/30/519630/7/check/openstack-tox-py36/e6ecd30/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/30/519630/7/check/openstack-tox-py37/311f41f/ : SUCCESS in 4m 41s\n- build-monasca-docker-image http://logs.openstack.org/30/519630/7/check/build-monasca-docker-image/e288749/ : SUCCESS in 6m 02s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/30/519630/7/check/monasca-tempest-python2-influxdb/dce3b8e/ : SUCCESS in 42m 37s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/30/519630/7/check/monasca-tempest-python3-influxdb/febf880/ : SUCCESS in 36m 13s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/30/519630/7/check/monasca-tempest-java-influxdb/cdbb243/ : FAILURE in 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"eda0c2f4d5577670d38f6db7249fcde0fbc8b913","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-10 10:09:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"59b437f74c13027500ae7bc01a84515ba42e1fef","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-10 10:30:36.000000000","message":"Patch Set 7:\n\nThanks for taking a look. That sounds fine to me regarding moving the docs.","accounts_in_message":[],"_revision_number":7},{"id":"ce0b2df93f184c0a6a332c3fa77a219cfa6bbbae","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-22 09:41:08.000000000","message":"Patch Set 7:\n\nWitek, would it be possible to follow up the doc refactor in a separate commit?","accounts_in_message":[],"_revision_number":7},{"id":"2657a34294d07f561ba58f5dc521de3565177464","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-29 11:47:26.000000000","message":"Patch Set 7:\n\nDoug, I\u0027m very sorry for the delay with these changes. I\u0027m totally fine with refactoring the docs later.\n\nWhat I would like to have though is to configure it in DevStack, so that the changes can be easily tested. As of now we still use the deprecated yaml configuration there.\n\nIdeally we could even add tempest tests for these plugins, but that could be added later.","accounts_in_message":[],"_revision_number":7},{"id":"68b458198529f4d78aaf6bfc4b5fc6c9e5574f5c","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-23 10:44:24.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d3359c31626ba091ee13ebca4685417dbaf30bfa","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-07 11:53:41.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLet\u0027s move forward with it.","accounts_in_message":[],"_revision_number":7},{"id":"927a8a9a68245d1bee058be3c1f3cc75d648f9e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 11:53:54.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9eee9a8371fe1b7e6d115a225b58b7e7215f31a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 12:04:50.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://logs.opendev.org/30/519630/7/gate/test-release-openstack/2cc8910/ : SUCCESS in 2m 39s\n- openstack-tox-lower-constraints https://logs.opendev.org/30/519630/7/gate/openstack-tox-lower-constraints/a11bb87/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://logs.opendev.org/30/519630/7/gate/openstack-tox-pep8/fa4c2b0/ : SUCCESS in 3m 41s\n- openstack-tox-py27 https://logs.opendev.org/30/519630/7/gate/openstack-tox-py27/66b049b/ : SUCCESS in 3m 11s\n- openstack-tox-py36 https://logs.opendev.org/30/519630/7/gate/openstack-tox-py36/640c907/ : SUCCESS in 5m 14s\n- openstack-tox-py37 https://logs.opendev.org/30/519630/7/gate/openstack-tox-py37/a6537ae/ : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":7},{"id":"f7fbf15e62f7f536c122cb1f01291b88bee8446a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 12:04:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"39a906b8fba280707e457052224210022407e140","revisions":{"cc07d60236a04061a345ed3ca1eaa66b5dd4839d":{"kind":"REWORK","_number":1,"created":"2017-11-14 12:09:55.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/30/519630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/30/519630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/1"}}},"commit":{"parents":[{"commit":"e13f529621c90eec7716053942a904c3190ba59c","subject":"Add .zuul.yaml file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e13f529621c90eec7716053942a904c3190ba59c"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 11:37:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 12:09:39.000000000","tz":0},"subject":"Templates for Slack notifications","message":"Templates for Slack notifications\n\nThis change adds an optional, user configurable template which\nmay be used to format the text contained in Slack notifications.\n\nImplements: blueprint templates-for-slack-notifications\nChange-Id: Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/cc07d60236a04061a345ed3ca1eaa66b5dd4839d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/cc07d60236a04061a345ed3ca1eaa66b5dd4839d"}]},"branch":"refs/heads/master"},"67e63112d940d46a742dac9eb9bdff58f166c7d1":{"kind":"REWORK","_number":2,"created":"2017-11-14 12:31:26.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/30/519630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/30/519630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/2"}}},"commit":{"parents":[{"commit":"e13f529621c90eec7716053942a904c3190ba59c","subject":"Add .zuul.yaml file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e13f529621c90eec7716053942a904c3190ba59c"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 11:37:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 12:31:14.000000000","tz":0},"subject":"Templates for Slack notifications","message":"Templates for Slack notifications\n\nThis change adds an optional, user configurable template which\nmay be used to format the text contained in Slack notifications.\n\nImplements: blueprint templates-for-slack-notifications\nChange-Id: Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/67e63112d940d46a742dac9eb9bdff58f166c7d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/67e63112d940d46a742dac9eb9bdff58f166c7d1"}]},"branch":"refs/heads/master"},"0ca4c1f4f111d10a0a18c92f55baae4584a4026e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2017-11-22 15:45:03.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/30/519630/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/30/519630/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/3"}}},"commit":{"parents":[{"commit":"e13f529621c90eec7716053942a904c3190ba59c","subject":"Add .zuul.yaml file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e13f529621c90eec7716053942a904c3190ba59c"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 11:37:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-22 15:44:20.000000000","tz":0},"subject":"Templates for Slack notifications","message":"Templates for Slack notifications\n\nThis change adds an optional, user configurable template which\nmay be used to format the text contained in Slack notifications.\n\nStory: 2001308\nChange-Id: Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/0ca4c1f4f111d10a0a18c92f55baae4584a4026e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/0ca4c1f4f111d10a0a18c92f55baae4584a4026e"}]},"branch":"refs/heads/master"},"b3a54534f03302e942213d8d524fd4b217fd6b51":{"kind":"REWORK","_number":4,"created":"2019-04-30 12:32:05.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/30/519630/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/30/519630/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/4"}}},"commit":{"parents":[{"commit":"ac837fa3cc96087c0b34a71665c2bd313613c7ce","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/ac837fa3cc96087c0b34a71665c2bd313613c7ce"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 11:37:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-30 12:31:53.000000000","tz":60},"subject":"Templates for Slack notifications","message":"Templates for Slack notifications\n\nThis change adds an optional, user configurable template which\nmay be used to format the text contained in Slack notifications.\n\nStory: 2001308\nChange-Id: Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/b3a54534f03302e942213d8d524fd4b217fd6b51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/b3a54534f03302e942213d8d524fd4b217fd6b51"}]},"branch":"refs/heads/master"},"82fe17787b64cf8dd0317a85f91fb5d6249fc447":{"kind":"REWORK","_number":5,"created":"2019-05-01 13:29:27.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/30/519630/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/30/519630/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/5"}}},"commit":{"parents":[{"commit":"ac837fa3cc96087c0b34a71665c2bd313613c7ce","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/ac837fa3cc96087c0b34a71665c2bd313613c7ce"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 11:37:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-05-01 13:29:12.000000000","tz":60},"subject":"Templates for Slack notifications","message":"Templates for Slack notifications\n\nThis change adds an optional, user configurable template which\nmay be used to format the text contained in Slack notifications.\n\nStory: 2001308\nChange-Id: Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/82fe17787b64cf8dd0317a85f91fb5d6249fc447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/82fe17787b64cf8dd0317a85f91fb5d6249fc447"}]},"branch":"refs/heads/master"},"4acca504c652f5704b20b1a55f2fbe57d3d7acb9":{"kind":"REWORK","_number":6,"created":"2019-05-01 14:42:45.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/30/519630/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/30/519630/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/6"}}},"commit":{"parents":[{"commit":"ac837fa3cc96087c0b34a71665c2bd313613c7ce","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/ac837fa3cc96087c0b34a71665c2bd313613c7ce"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 11:37:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-05-01 14:42:30.000000000","tz":60},"subject":"Templates for Slack notifications","message":"Templates for Slack notifications\n\nThis change adds an optional, user configurable template which\nmay be used to format the text contained in Slack notifications.\n\nStory: 2001308\nChange-Id: Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/4acca504c652f5704b20b1a55f2fbe57d3d7acb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/4acca504c652f5704b20b1a55f2fbe57d3d7acb9"}]},"branch":"refs/heads/master"},"39a906b8fba280707e457052224210022407e140":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-05-02 08:53:17.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/30/519630/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/30/519630/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/30/519630/7"}}},"commit":{"parents":[{"commit":"ac837fa3cc96087c0b34a71665c2bd313613c7ce","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/ac837fa3cc96087c0b34a71665c2bd313613c7ce"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2017-11-14 11:37:09.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-05-02 08:49:42.000000000","tz":60},"subject":"Templates for Slack notifications","message":"Templates for Slack notifications\n\nThis change adds an optional, user configurable template which\nmay be used to format the text contained in Slack notifications.\n\nStory: 2001308\nTask: 5859\nChange-Id: Id936c3dc8b4f3e2430de20c8b69d0e703b1cf9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/39a906b8fba280707e457052224210022407e140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/39a906b8fba280707e457052224210022407e140"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
