)]}'
{"id":"openstack%2Fmonasca-notification~663960","triplet_id":"openstack%2Fmonasca-notification~master~I99d7750fac6d0355d182df1612c6a5abbbad7fc6","project":"openstack/monasca-notification","branch":"master","topic":"plugin-detection-loop-fix","hashtags":[],"change_id":"I99d7750fac6d0355d182df1612c6a5abbbad7fc6","subject":"Fix notification type plugin install loop","status":"MERGED","created":"2019-06-07 14:32:07.000000000","updated":"2019-07-03 16:50:44.000000000","submitted":"2019-07-03 16:50:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663960-1562172644900-b16e9fe7","meta_rev_id":"b217114485736c9358975ffb2e724c2a58a64665","_number":663960,"virtual_id_number":663960,"owner":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-03 16:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-03 16:37:21.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-02 11:48:30.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-03 16:37:21.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 11:48:30.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-07-03 16:37:21.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-03 16:50:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6e6a03e4e8be36f32b2be565a2ff337f18c4ff1","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-06-07 14:32:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1142d028adb169de9726dec0e29ff0ba0996c958","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-06-07 14:33:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eeca70def01487d689cbe41f0d9bb0aa1574a170","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-06-07 14:40:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"43bceb2f1705540995cf6bb72b3501aed53f6f0d","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-06-07 14:42:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f760432143a079f33d93ab4f2c3ddcfd0b7f4b3e","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-06-07 14:44:36.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6cfabbccd5f2bc841fed1f16821165c48de8a050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:37:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/60/663960/5/check/openstack-tox-cover/c7aa175/cover/ : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/663960/5/check/openstack-tox-lower-constraints/e1b1b62/ : SUCCESS in 5m 17s\n- openstack-tox-pep8 http://logs.openstack.org/60/663960/5/check/openstack-tox-pep8/ec8cce5/ : FAILURE in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/60/663960/5/check/openstack-tox-py27/8cbb65c/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/60/663960/5/check/openstack-tox-py36/c702ea2/ : SUCCESS in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/60/663960/5/check/openstack-tox-py37/e6924ce/ : SUCCESS in 9m 13s\n- build-monasca-docker-image http://logs.openstack.org/60/663960/5/check/build-monasca-docker-image/a65390d/ : SUCCESS in 6m 36s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/60/663960/5/check/monasca-tempest-python2-influxdb/a42ee01/ : FAILURE in 33m 53s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/60/663960/5/check/monasca-tempest-python3-influxdb/3a3eee2/ : FAILURE in 34m 12s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/60/663960/5/check/monasca-tempest-java-influxdb/206b08d/ : FAILURE in 49m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d37b92253f8f29762524961e53e1e2501fa60784","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-06-07 15:49:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b6ec1ba121a9be364312c6294a9803a80a66611e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 16:47:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/60/663960/6/check/openstack-tox-cover/d259588/cover/ : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/663960/6/check/openstack-tox-lower-constraints/6f8a62e/ : SUCCESS in 6m 07s\n- openstack-tox-pep8 http://logs.openstack.org/60/663960/6/check/openstack-tox-pep8/4236e0a/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/60/663960/6/check/openstack-tox-py27/0e44afa/ : SUCCESS in 7m 29s\n- openstack-tox-py36 http://logs.openstack.org/60/663960/6/check/openstack-tox-py36/633f50a/ : SUCCESS in 8m 06s\n- openstack-tox-py37 http://logs.openstack.org/60/663960/6/check/openstack-tox-py37/8c0307a/ : SUCCESS in 9m 22s\n- build-monasca-docker-image http://logs.openstack.org/60/663960/6/check/build-monasca-docker-image/9c269da/ : SUCCESS in 7m 00s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/60/663960/6/check/monasca-tempest-python2-influxdb/db3a4e7/ : FAILURE in 40m 50s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/60/663960/6/check/monasca-tempest-python3-influxdb/b6b6663/ : SUCCESS in 50m 11s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/60/663960/6/check/monasca-tempest-java-influxdb/cf2605a/ : FAILURE in 44m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a8f65e02d53660b1973915fa4b6f0fb07e1c88e2","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-02 10:27:43.000000000","message":"Patch Set 6:\n\nThanks, tested in DevStack and works as expected.\n\nI added this change as dependency for https://review.opendev.org/633194 to test it in CI as well. Waiting for the checks to finish.","accounts_in_message":[],"_revision_number":6},{"id":"261f39da7a8e0de7162283006404c164eada6d8d","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-02 11:48:30.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4a6fc404f62330036ffd90c5bfcc739643a6bcb8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-03 16:37:21.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"72bab871da4e17014e3a2eba6c381f22a48b80ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:37:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4ce444942d771ab2170b33b65cd69c58a9dc12de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:50:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/60/663960/6/gate/openstack-tox-lower-constraints/7d888f0/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/60/663960/6/gate/openstack-tox-pep8/c49631b/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/60/663960/6/gate/openstack-tox-py27/8e01302/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/60/663960/6/gate/openstack-tox-py36/854e1ac/ : SUCCESS in 2m 54s\n- openstack-tox-py37 http://logs.openstack.org/60/663960/6/gate/openstack-tox-py37/791672d/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"1a1bdca8afa40c9dad079b499191e3153f921875","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:50:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"df2f1ef228c0d33cc9f8c34f80221f2ac624e3ca","revisions":{"b9616ba4a45b71483a574ca044d7bd93fd658458":{"kind":"REWORK","_number":1,"created":"2019-06-07 14:32:07.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/60/663960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/60/663960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/1"}}},"commit":{"parents":[{"commit":"f708d32bf19dbfd843a6a6f0a05667400856e490","subject":"Change repository links usage in Dockerfile and build image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f708d32bf19dbfd843a6a6f0a05667400856e490"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:27:56.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:32:05.000000000","tz":60},"subject":"Fix notification type plugin install loop","message":"Fix notification type plugin install loop\n\nThe \u0027insert_configured_plugins\u0027 function occasionally gets into an infinite loop.\nThis patch ensures the loop fails properly after max_retries is met.\n\nChange-Id: I99d7750fac6d0355d182df1612c6a5abbbad7fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/b9616ba4a45b71483a574ca044d7bd93fd658458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/b9616ba4a45b71483a574ca044d7bd93fd658458"}]},"branch":"refs/heads/master"},"53ade557a184d10ce9c9217ab61ae85fdeaf78c2":{"kind":"REWORK","_number":2,"created":"2019-06-07 14:33:07.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/60/663960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/60/663960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/2"}}},"commit":{"parents":[{"commit":"f708d32bf19dbfd843a6a6f0a05667400856e490","subject":"Change repository links usage in Dockerfile and build image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f708d32bf19dbfd843a6a6f0a05667400856e490"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:27:56.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:33:03.000000000","tz":0},"subject":"Fix notification type plugin install loop","message":"Fix notification type plugin install loop\n\nThe \u0027insert_configured_plugins\u0027 function occasionally gets into an infinite loop.\nThis patch ensures the loop fails properly after max_retries is met.\n\nChange-Id: I99d7750fac6d0355d182df1612c6a5abbbad7fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/53ade557a184d10ce9c9217ab61ae85fdeaf78c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/53ade557a184d10ce9c9217ab61ae85fdeaf78c2"}]},"branch":"refs/heads/master"},"190bd95b6e197c7023cb527e7fdc6f3530537db3":{"kind":"REWORK","_number":3,"created":"2019-06-07 14:40:18.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/60/663960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/60/663960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/3"}}},"commit":{"parents":[{"commit":"f708d32bf19dbfd843a6a6f0a05667400856e490","subject":"Change repository links usage in Dockerfile and build image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f708d32bf19dbfd843a6a6f0a05667400856e490"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:27:56.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:39:34.000000000","tz":60},"subject":"Fix notification type plugin install loop","message":"Fix notification type plugin install loop\n\nThe \u0027insert_configured_plugins\u0027 function occasionally gets into an infinite loop.\nThis patch ensures the loop fails properly after max_retries is met.\n\nStory: 2005540\nTask: 30680\nChange-Id: I99d7750fac6d0355d182df1612c6a5abbbad7fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/190bd95b6e197c7023cb527e7fdc6f3530537db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/190bd95b6e197c7023cb527e7fdc6f3530537db3"}]},"branch":"refs/heads/master"},"0d38f59097943ca9846cb981345ff028b3cfaa20":{"kind":"REWORK","_number":4,"created":"2019-06-07 14:42:06.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/60/663960/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/60/663960/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/4"}}},"commit":{"parents":[{"commit":"f708d32bf19dbfd843a6a6f0a05667400856e490","subject":"Change repository links usage in Dockerfile and build image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f708d32bf19dbfd843a6a6f0a05667400856e490"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:27:56.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:41:57.000000000","tz":60},"subject":"Fix notification type plugin install loop","message":"Fix notification type plugin install loop\n\nThe \u0027insert_configured_plugins\u0027 function occasionally gets into an infinite loop.\nThis patch ensures the loop fails properly after max_retries is met.\n\nStory: 2005540\nTask: 30680\nChange-Id: I99d7750fac6d0355d182df1612c6a5abbbad7fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/0d38f59097943ca9846cb981345ff028b3cfaa20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/0d38f59097943ca9846cb981345ff028b3cfaa20"}]},"branch":"refs/heads/master"},"e9a75059c0904ab3954cbd7a29aaaf1905705052":{"kind":"NO_CHANGE","_number":5,"created":"2019-06-07 14:44:36.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/60/663960/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/60/663960/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/5"}}},"commit":{"parents":[{"commit":"f708d32bf19dbfd843a6a6f0a05667400856e490","subject":"Change repository links usage in Dockerfile and build image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f708d32bf19dbfd843a6a6f0a05667400856e490"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:27:56.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:44:27.000000000","tz":60},"subject":"Fix notification type plugin install loop","message":"Fix notification type plugin install loop\n\nThe \u0027insert_configured_plugins\u0027 function occasionally gets into an infinite loop.\nThis patch ensures the loop fails properly after max_retries is met.\n\nStory: 2005540\nTask: 30680\nChange-Id: I99d7750fac6d0355d182df1612c6a5abbbad7fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e9a75059c0904ab3954cbd7a29aaaf1905705052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e9a75059c0904ab3954cbd7a29aaaf1905705052"}]},"branch":"refs/heads/master"},"df2f1ef228c0d33cc9f8c34f80221f2ac624e3ca":{"kind":"REWORK","_number":6,"created":"2019-06-07 15:49:36.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/60/663960/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/60/663960/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/60/663960/6"}}},"commit":{"parents":[{"commit":"f708d32bf19dbfd843a6a6f0a05667400856e490","subject":"Change repository links usage in Dockerfile and build image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f708d32bf19dbfd843a6a6f0a05667400856e490"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 14:27:56.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-06-07 15:49:27.000000000","tz":60},"subject":"Fix notification type plugin install loop","message":"Fix notification type plugin install loop\n\nThe \u0027insert_configured_plugins\u0027 function occasionally gets into an infinite loop.\nThis patch ensures the loop fails properly after max_retries is met.\n\nStory: 2005540\nTask: 30680\nChange-Id: I99d7750fac6d0355d182df1612c6a5abbbad7fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/df2f1ef228c0d33cc9f8c34f80221f2ac624e3ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/df2f1ef228c0d33cc9f8c34f80221f2ac624e3ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
