)]}'
{"id":"openstack%2Fmonasca-notification~667920","triplet_id":"openstack%2Fmonasca-notification~master~Id1f963b987f36a970ada738fc56e23000430eef1","project":"openstack/monasca-notification","branch":"master","topic":"bindep-fallback","hashtags":[],"change_id":"Id1f963b987f36a970ada738fc56e23000430eef1","subject":"Add local bindep.txt","status":"ABANDONED","created":"2019-06-27 12:44:25.000000000","updated":"2019-06-27 16:02:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fefeb292874b45901f61571274d45473042721ab","_number":667920,"virtual_id_number":667920,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2019-06-27 15:53:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 13:44:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-27 15:53:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c18b831a9cf9356cb58a0cfa61d5942b554eb879","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-27 12:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37dce0515bc252e8fe70e118d63ab189e1de32db","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-27 13:26:15.000000000","message":"Topic set to bindep-fallback","accounts_in_message":[],"_revision_number":1},{"id":"1ce348f14374865b5dc90ce4cf4753b9c66afbda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:44:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/20/667920/1/check/openstack-tox-cover/9c6c75e/ : RETRY_LIMIT in 2m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667920/1/check/openstack-tox-lower-constraints/0420c2d/ : RETRY_LIMIT in 2m 06s\n- openstack-tox-pep8 http://logs.openstack.org/20/667920/1/check/openstack-tox-pep8/2e9cbc1/ : RETRY_LIMIT in 1m 58s\n- openstack-tox-py27 http://logs.openstack.org/20/667920/1/check/openstack-tox-py27/3b883dc/ : RETRY_LIMIT in 2m 00s\n- openstack-tox-py36 http://logs.openstack.org/20/667920/1/check/openstack-tox-py36/3208fff/ : RETRY_LIMIT in 2m 19s\n- openstack-tox-py37 http://logs.openstack.org/20/667920/1/check/openstack-tox-py37/4e657ee/ : RETRY_LIMIT in 1m 58s\n- build-openstack-releasenotes http://logs.openstack.org/20/667920/1/check/build-openstack-releasenotes/b42ab1b/ : RETRY_LIMIT in 1m 49s\n- build-monasca-docker-image http://logs.openstack.org/20/667920/1/check/build-monasca-docker-image/6507ce0/ : SUCCESS in 6m 02s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/20/667920/1/check/monasca-tempest-python2-influxdb/6bae2a1/ : FAILURE in 55m 12s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/20/667920/1/check/monasca-tempest-python3-influxdb/99d9ab2/ : SUCCESS in 38m 37s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/20/667920/1/check/monasca-tempest-java-influxdb/0976b74/ : FAILURE in 49m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9551df60d48a7a6ea5a535df38273f916b681573","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-27 13:56:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5650158adb3ca63153dd794d6a7111bb11dd24d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:52:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/20/667920/1/check/openstack-tox-cover/b49095a/ : RETRY_LIMIT in 2m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667920/1/check/openstack-tox-lower-constraints/552e130/ : RETRY_LIMIT in 2m 14s\n- openstack-tox-pep8 http://logs.openstack.org/20/667920/1/check/openstack-tox-pep8/b067843/ : RETRY_LIMIT in 1m 54s\n- openstack-tox-py27 http://logs.openstack.org/20/667920/1/check/openstack-tox-py27/78b56a7/ : RETRY_LIMIT in 2m 33s\n- openstack-tox-py36 http://logs.openstack.org/20/667920/1/check/openstack-tox-py36/b03a8e2/ : RETRY_LIMIT in 2m 51s\n- openstack-tox-py37 http://logs.openstack.org/20/667920/1/check/openstack-tox-py37/62e5883/ : RETRY_LIMIT in 2m 14s\n- build-openstack-releasenotes http://logs.openstack.org/20/667920/1/check/build-openstack-releasenotes/f8817a8/ : RETRY_LIMIT in 2m 13s\n- build-monasca-docker-image http://logs.openstack.org/20/667920/1/check/build-monasca-docker-image/26fe272/ : SUCCESS in 6m 10s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/20/667920/1/check/monasca-tempest-python2-influxdb/d53c00f/ : SUCCESS in 42m 07s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/20/667920/1/check/monasca-tempest-python3-influxdb/e50c7b8/ : SUCCESS in 37m 58s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/20/667920/1/check/monasca-tempest-java-influxdb/b71b1a9/ : FAILURE in 46m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c363d38fd5aac04189d9597d4f14e91117ec4c4","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-27 15:37:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e39ffd29ad4ec2c95e2c1338cc640a847fc3a591","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 15:53:33.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good and the previously failing tox-py37 runs successfully.","accounts_in_message":[],"_revision_number":2},{"id":"fefeb292874b45901f61571274d45473042721ab","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-27 16:02:52.000000000","message":"Abandoned\n\nCode added to https://review.opendev.org/667585","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"31aeb488aa0be17469953f61a05dd1aa28b3d776","revisions":{"fc0a7e148b15d4a82578738f19fd85091448f6da":{"kind":"REWORK","_number":1,"created":"2019-06-27 12:44:25.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/20/667920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/20/667920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/20/667920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/20/667920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/20/667920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/20/667920/1"}}},"commit":{"parents":[{"commit":"f708d32bf19dbfd843a6a6f0a05667400856e490","subject":"Change repository links usage in Dockerfile and build image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f708d32bf19dbfd843a6a6f0a05667400856e490"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-27 11:56:32.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-27 11:56:32.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nThe file includes PostgreSQL dependendencies for testing.\n\nChange-Id: Id1f963b987f36a970ada738fc56e23000430eef1\nStory: 2006058\nTask: 34771\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/fc0a7e148b15d4a82578738f19fd85091448f6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/fc0a7e148b15d4a82578738f19fd85091448f6da"}]},"branch":"refs/heads/master"},"31aeb488aa0be17469953f61a05dd1aa28b3d776":{"kind":"REWORK","_number":2,"created":"2019-06-27 15:37:48.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/20/667920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/20/667920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/20/667920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/20/667920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/20/667920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/20/667920/2"}}},"commit":{"parents":[{"commit":"f708d32bf19dbfd843a6a6f0a05667400856e490","subject":"Change repository links usage in Dockerfile and build image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f708d32bf19dbfd843a6a6f0a05667400856e490"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-27 11:56:32.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-06-27 15:37:36.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nThe file includes PostgreSQL dependendencies for testing.\n\nChange-Id: Id1f963b987f36a970ada738fc56e23000430eef1\nStory: 2006058\nTask: 34771\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/31aeb488aa0be17469953f61a05dd1aa28b3d776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/31aeb488aa0be17469953f61a05dd1aa28b3d776"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
