)]}'
{"id":"openstack%2Fmonasca-notification~668508","triplet_id":"openstack%2Fmonasca-notification~master~Icb91feaf7821fe014542350125c5011bab530103","project":"openstack/monasca-notification","branch":"master","hashtags":[],"change_id":"Icb91feaf7821fe014542350125c5011bab530103","subject":"Add default notification plugins to configuration","status":"MERGED","created":"2019-07-01 16:01:19.000000000","updated":"2019-07-02 09:28:29.000000000","submitted":"2019-07-02 09:28:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668508-1562059709757-e6a449f5","meta_rev_id":"53003b2b808e92bc47309cba2429e404ce2d7bc2","_number":668508,"virtual_id_number":668508,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-02 09:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":2,"date":"2019-07-01 16:15:48.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":1,"date":"2019-07-02 09:17:25.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 16:15:48.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-02 09:17:25.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-07-02 09:28:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cae75c3b5950056aa47a79b39e1d9f2c6e4a771a","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-01 16:01:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3170e5df87783c2a2eb3cfd348b18cbb986de8bf","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-01 16:15:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"821b7063874f2e5ecfa3e60f71026fddec1b1ada","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:59:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/08/668508/1/check/openstack-tox-cover/95daf3c/cover/ : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/668508/1/check/openstack-tox-lower-constraints/4e67d56/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/08/668508/1/check/openstack-tox-pep8/9d51ca1/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/08/668508/1/check/openstack-tox-py27/6da04a6/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/08/668508/1/check/openstack-tox-py36/c3d5083/ : SUCCESS in 3m 16s\n- openstack-tox-py37 http://logs.openstack.org/08/668508/1/check/openstack-tox-py37/9b74a8a/ : SUCCESS in 4m 14s\n- build-monasca-docker-image http://logs.openstack.org/08/668508/1/check/build-monasca-docker-image/3bc490d/ : SUCCESS in 6m 23s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/08/668508/1/check/monasca-tempest-python2-influxdb/8a64d2e/ : SUCCESS in 39m 49s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/08/668508/1/check/monasca-tempest-python3-influxdb/7a3cfa3/ : SUCCESS in 35m 29s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/08/668508/1/check/monasca-tempest-java-influxdb/1113018/ : FAILURE in 47m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae63cf189daa3efc381c2f0a4cc7b859368d4f2e","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-07-02 09:17:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc096e0fde0c13be2c3b85d6ddaafbe21d3e46bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 09:17:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09b8d16ef0f6b57be0231732ab7848697db05174","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 09:28:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/08/668508/1/gate/openstack-tox-lower-constraints/0eb4363/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/08/668508/1/gate/openstack-tox-pep8/23fb802/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/08/668508/1/gate/openstack-tox-py27/0f001ed/ : SUCCESS in 2m 51s\n- openstack-tox-py36 http://logs.openstack.org/08/668508/1/gate/openstack-tox-py36/0a584ae/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/08/668508/1/gate/openstack-tox-py37/e5e96bc/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"53003b2b808e92bc47309cba2429e404ce2d7bc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 09:28:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd1c6dfab9719b4252da29789aeb1f7974419e54","revisions":{"fd1c6dfab9719b4252da29789aeb1f7974419e54":{"kind":"REWORK","_number":1,"created":"2019-07-01 16:01:19.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/08/668508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/08/668508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/08/668508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/08/668508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/08/668508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/08/668508/1"}}},"commit":{"parents":[{"commit":"07da3274dcde1fcc54f9216e646dd9c2896d0971","subject":"Document missing Grafana url env var in Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/07da3274dcde1fcc54f9216e646dd9c2896d0971"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-01 15:46:51.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-01 15:46:51.000000000","tz":120},"subject":"Add default notification plugins to configuration","message":"Add default notification plugins to configuration\n\nIn the past email, pagerduty and webhook notification plugins had been\nhard coded and always enabled. This situation is still reflected in the\ndeprecated YAML configuration file [1]. These notification method types\nare also expected by tempest tests.\n\nThis change sets email, pagerduty and webhook as the default list of\nenabled notifiers. This corresponds to the old behavior and testing but\ncan be easily modified by overwriting the list in the configuration\nfile.\n\n[1] https://opendev.org/openstack/monasca-api/src/branch/master/devstack/files/monasca-notification/notification.yaml#L51-L54\n\nChange-Id: Icb91feaf7821fe014542350125c5011bab530103\nStory: 2000959\nTask: 35644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/fd1c6dfab9719b4252da29789aeb1f7974419e54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/fd1c6dfab9719b4252da29789aeb1f7974419e54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
