)]}'
{"id":"openstack%2Fmonasca-notification~669410","triplet_id":"openstack%2Fmonasca-notification~master~I87be8b97790f73256e9b26b9756b5e8a2461314d","project":"openstack/monasca-notification","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I87be8b97790f73256e9b26b9756b5e8a2461314d","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 17:18:31.000000000","updated":"2019-07-08 09:37:58.000000000","submitted":"2019-07-08 09:37:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669410-1562578678830-7de5f3e4","meta_rev_id":"23defab6f68af115bc50c6b35fe2a85b1f212a58","_number":669410,"virtual_id_number":669410,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-08 09:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-08 09:27:39.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-08 09:27:39.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 09:27:39.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-07-08 09:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d29e7f1a963e72973480f1422a0b0d36ad44260c","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 17:18:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"052b58315690c0211de38db1f566836a60e25e9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 18:11:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/10/669410/1/check/openstack-tox-cover/7e95af3/cover/ : SUCCESS in 3m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/669410/1/check/openstack-tox-lower-constraints/9b4e15f/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/10/669410/1/check/openstack-tox-pep8/555f8dd/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/10/669410/1/check/openstack-tox-py27/652a686/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/10/669410/1/check/openstack-tox-py36/1db1f87/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/10/669410/1/check/openstack-tox-py37/3201944/ : SUCCESS in 3m 39s\n- build-openstack-releasenotes http://logs.openstack.org/10/669410/1/check/build-openstack-releasenotes/35179a3/html/ : SUCCESS in 2m 45s\n- build-monasca-docker-image http://logs.openstack.org/10/669410/1/check/build-monasca-docker-image/c1aea4b/ : SUCCESS in 7m 35s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/10/669410/1/check/monasca-tempest-python2-influxdb/2ed4924/ : SUCCESS in 35m 26s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/10/669410/1/check/monasca-tempest-python3-influxdb/c2c072f/ : SUCCESS in 39m 45s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/10/669410/1/check/monasca-tempest-java-influxdb/23d10f2/ : FAILURE in 50m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27d1f7ea8a9b2a3e34268ddbfa3ecae21b0ef3e2","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-08 09:27:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"432b1c41411f4a2b96fd14e9dc931e2a65a61ac4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 09:27:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0eb296d7d2f48c63c0e49adb6ae5ce4da6a25391","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 09:37:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5f4ddc92b412380930d031bd083ab1f93b7e003","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 09:37:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/10/669410/1/gate/openstack-tox-lower-constraints/c0b8fef/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/10/669410/1/gate/openstack-tox-pep8/904ef18/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/10/669410/1/gate/openstack-tox-py27/58815fa/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/10/669410/1/gate/openstack-tox-py36/9e9ae2f/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/10/669410/1/gate/openstack-tox-py37/2da1bf2/ : SUCCESS in 4m 59s\n- build-openstack-releasenotes http://logs.openstack.org/10/669410/1/gate/build-openstack-releasenotes/7da65c4/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd304f3ac3363714952d12a13cb30ba2164a7fff","revisions":{"cd304f3ac3363714952d12a13cb30ba2164a7fff":{"kind":"REWORK","_number":1,"created":"2019-07-05 17:18:31.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/10/669410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/10/669410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/669410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/669410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/10/669410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/10/669410/1"}}},"commit":{"parents":[{"commit":"bcbd8d08de2db6388867271298b7a604f8b486ca","subject":"Merge \"Fix notification type plugin install loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/bcbd8d08de2db6388867271298b7a604f8b486ca"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 17:18:08.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 17:18:30.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I87be8b97790f73256e9b26b9756b5e8a2461314d\nStory: #2005924\nTask: #34223\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/cd304f3ac3363714952d12a13cb30ba2164a7fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/cd304f3ac3363714952d12a13cb30ba2164a7fff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
