)]}'
{"id":"openstack%2Fmonasca-notification~670117","triplet_id":"openstack%2Fmonasca-notification~master~Ie0ddefd24a8e53246fce53c3ca166af9f50d844d","project":"openstack/monasca-notification","branch":"master","hashtags":[],"change_id":"Ie0ddefd24a8e53246fce53c3ca166af9f50d844d","subject":"Disable monasca-tempest-java-influxdb job","status":"MERGED","created":"2019-07-10 16:21:28.000000000","updated":"2019-07-15 08:34:32.000000000","submitted":"2019-07-15 08:34:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670117-1563179672588-e25996b4","meta_rev_id":"b6b0d302a3c7354a7242ce9d039f779c06e47146","_number":670117,"virtual_id_number":670117,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-15 08:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-15 08:24:51.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-15 08:25:06.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 08:25:06.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-07-15 08:34:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"937c0781a0efa055f7ba67dcd3cc8a5d9bba3f4b","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-10 16:21:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1b7f1beecec6d5d2b12796590280529c8f19f06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:07:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/17/670117/1/check/openstack-tox-cover/e118e8d/cover/ : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/670117/1/check/openstack-tox-lower-constraints/beb3c5e/ : SUCCESS in 5m 52s\n- openstack-tox-pep8 http://logs.openstack.org/17/670117/1/check/openstack-tox-pep8/87f8b22/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/17/670117/1/check/openstack-tox-py27/cf8f4ff/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/17/670117/1/check/openstack-tox-py36/76d68e1/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/17/670117/1/check/openstack-tox-py37/d25a5ef/ : SUCCESS in 5m 47s\n- build-monasca-docker-image http://logs.openstack.org/17/670117/1/check/build-monasca-docker-image/b9d9566/ : SUCCESS in 7m 31s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/17/670117/1/check/monasca-tempest-python2-influxdb/5459d95/ : SUCCESS in 34m 56s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/17/670117/1/check/monasca-tempest-python3-influxdb/bf7c248/ : SUCCESS in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e8c8f44213d4e77b3bc086aed469ac0cc88b758","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-15 08:24:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"43e16ca44be41d3051a5f452f1251193895b5db0","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-15 08:25:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c5489be9a9dee5f1b70464048a96956c3bd288a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:25:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab30187eee9ad3a8341a94b43da049ce68deaf78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:34:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b6b0d302a3c7354a7242ce9d039f779c06e47146","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:34:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/17/670117/1/gate/openstack-tox-lower-constraints/e7bd826/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/17/670117/1/gate/openstack-tox-pep8/2ac449c/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/17/670117/1/gate/openstack-tox-py27/9187591/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/17/670117/1/gate/openstack-tox-py36/1103fbe/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/17/670117/1/gate/openstack-tox-py37/dbd1bc6/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04b10032b783081303af0431574a805d59bdc2dc","revisions":{"04b10032b783081303af0431574a805d59bdc2dc":{"kind":"REWORK","_number":1,"created":"2019-07-10 16:21:28.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/17/670117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/17/670117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/17/670117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/17/670117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/17/670117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/17/670117/1"}}},"commit":{"parents":[{"commit":"cd304f3ac3363714952d12a13cb30ba2164a7fff","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/cd304f3ac3363714952d12a13cb30ba2164a7fff"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-10 16:20:45.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-10 16:20:45.000000000","tz":120},"subject":"Disable monasca-tempest-java-influxdb job","message":"Disable monasca-tempest-java-influxdb job\n\nThe job has been failing for a long time. Java implementation of\nInfluxDB persister is not maintained.\n\nChange-Id: Ie0ddefd24a8e53246fce53c3ca166af9f50d844d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/04b10032b783081303af0431574a805d59bdc2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/04b10032b783081303af0431574a805d59bdc2dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
