)]}'
{"id":"openstack%2Fmonasca-notification~672915","triplet_id":"openstack%2Fmonasca-notification~master~Ib1dba51a57fdce028a6b20da06201f0597085047","project":"openstack/monasca-notification","branch":"master","topic":"fix-python-json-bug","hashtags":[],"change_id":"Ib1dba51a57fdce028a6b20da06201f0597085047","subject":"Json loads error","status":"MERGED","created":"2019-07-26 08:41:48.000000000","updated":"2019-07-30 15:38:52.000000000","submitted":"2019-07-30 15:38:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672915-1564501132902-520616fe","meta_rev_id":"1756cbc2b2a90c7fa87455ffef02a6505787dcf8","_number":672915,"virtual_id_number":672915,"owner":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-30 15:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-30 15:00:33.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-30 15:28:00.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-30 15:28:00.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 15:00:33.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-30 15:28:00.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-07-30 15:38:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5cfa6afa3e4d975b743becd7a9f78e6371214da","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-07-26 08:41:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3367f610483548720bbbbdccd5288ae460ba95e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:30:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/15/672915/1/check/openstack-tox-cover/ed9812c/cover/ : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/672915/1/check/openstack-tox-lower-constraints/02ca413/ : SUCCESS in 5m 15s\n- openstack-tox-pep8 http://logs.openstack.org/15/672915/1/check/openstack-tox-pep8/b6918af/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/15/672915/1/check/openstack-tox-py27/7ae2694/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/15/672915/1/check/openstack-tox-py36/b4a3934/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/15/672915/1/check/openstack-tox-py37/7901fa1/ : SUCCESS in 4m 38s\n- build-monasca-docker-image http://logs.openstack.org/15/672915/1/check/build-monasca-docker-image/1088ab6/ : SUCCESS in 7m 14s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/15/672915/1/check/monasca-tempest-python2-influxdb/f0e088a/ : SUCCESS in 44m 39s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/15/672915/1/check/monasca-tempest-python3-influxdb/971cbb5/ : FAILURE in 45m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87448b6223c9e0d8cb9169721578b56cc7d57284","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-29 07:25:28.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for catching this. One comment inline.","accounts_in_message":[],"_revision_number":1},{"id":"c103f97ff52fc0ef7e29fed4ac3f4e5062f599b7","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-07-29 10:12:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8750e6bda2041385d4272a3fecd5fbde4372f7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 10:57:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/15/672915/2/check/requirements-check/a60fcdb/ : FAILURE in 3m 28s\n- openstack-tox-cover http://logs.openstack.org/15/672915/2/check/openstack-tox-cover/e21abf3/cover/ : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/672915/2/check/openstack-tox-lower-constraints/b3b2b4f/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/15/672915/2/check/openstack-tox-pep8/4ff3f8c/ : FAILURE in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/15/672915/2/check/openstack-tox-py27/bae3e51/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/15/672915/2/check/openstack-tox-py36/1db6cb6/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/15/672915/2/check/openstack-tox-py37/ae6f29e/ : SUCCESS in 4m 58s\n- build-monasca-docker-image http://logs.openstack.org/15/672915/2/check/build-monasca-docker-image/9478ca0/ : SUCCESS in 7m 01s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/15/672915/2/check/monasca-tempest-python2-influxdb/fc21b7d/ : FAILURE in 41m 52s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/15/672915/2/check/monasca-tempest-python3-influxdb/f23902a/ : FAILURE in 39m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f39e23585ee110e78a0ba8d1bc288c39e29cc08","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-29 11:23:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2de5e95d9e42d046893f8e74bf5bcdc29e0ca441","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-07-30 00:23:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ef9a0352c4ae4d9f7c05709d2af1eee309efed0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 01:06:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/15/672915/3/check/requirements-check/faf5b53/ : SUCCESS in 2m 29s\n- openstack-tox-cover http://logs.openstack.org/15/672915/3/check/openstack-tox-cover/c8ed2cc/cover/ : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/672915/3/check/openstack-tox-lower-constraints/b5aa6a8/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/15/672915/3/check/openstack-tox-pep8/aaf28ec/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/15/672915/3/check/openstack-tox-py27/5c0ff87/ : SUCCESS in 3m 01s\n- openstack-tox-py36 http://logs.openstack.org/15/672915/3/check/openstack-tox-py36/6459e58/ : SUCCESS in 3m 03s\n- openstack-tox-py37 http://logs.openstack.org/15/672915/3/check/openstack-tox-py37/9f07cc5/ : SUCCESS in 4m 14s\n- build-monasca-docker-image http://logs.openstack.org/15/672915/3/check/build-monasca-docker-image/19161fe/ : SUCCESS in 6m 25s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/15/672915/3/check/monasca-tempest-python2-influxdb/6179c24/ : SUCCESS in 37m 30s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/15/672915/3/check/monasca-tempest-python3-influxdb/e5a5520/ : SUCCESS in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c76c4fe14b038ad98781ba5b339050793c503ccf","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-30 15:00:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d8769ef518be49c0373301369c1d0289ccf813d1","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-30 15:28:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f172fe557beace888d4d3026f2690d1c4624d552","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 15:28:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6529d96b6d10d7f701dcc959ecf17475afcbdb47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 15:38:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/15/672915/3/gate/requirements-check/f00c3a8/ : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/672915/3/gate/openstack-tox-lower-constraints/64ac17d/ : SUCCESS in 6m 30s\n- openstack-tox-pep8 http://logs.openstack.org/15/672915/3/gate/openstack-tox-pep8/3b7f73a/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/15/672915/3/gate/openstack-tox-py27/247e270/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/15/672915/3/gate/openstack-tox-py36/6909833/ : SUCCESS in 6m 26s\n- openstack-tox-py37 http://logs.openstack.org/15/672915/3/gate/openstack-tox-py37/3fba4eb/ : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":3},{"id":"90dac64afeba0a149db72c5c1cf34b8210d03dd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 15:38:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"22d7c9bb514c9199fce45440ea4b452721ce2a96","revisions":{"f11457ce18c4b7958e633175459563eeb98e1163":{"kind":"REWORK","_number":1,"created":"2019-07-26 08:41:48.000000000","uploader":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"ref":"refs/changes/15/672915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/15/672915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/1"}}},"commit":{"parents":[{"commit":"04b10032b783081303af0431574a805d59bdc2dc","subject":"Disable monasca-tempest-java-influxdb job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/04b10032b783081303af0431574a805d59bdc2dc"}]}],"author":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-07-26 08:35:53.000000000","tz":480},"committer":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-07-26 08:40:29.000000000","tz":480},"subject":"Json loads error","message":"Json loads error\n\nPyhton version below 3.6,function json.loads() throw decoder error.\n3.6 version input option can bytes or bytearray,below 3.6 version\ntype of bytes not cimpatible.\n\nChange-Id: Ib1dba51a57fdce028a6b20da06201f0597085047\nTask: 35991\nStory: 2006285\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f11457ce18c4b7958e633175459563eeb98e1163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f11457ce18c4b7958e633175459563eeb98e1163"}]},"branch":"refs/heads/master"},"0f96fd652cf4fcb97360249c947337313f00ec77":{"kind":"REWORK","_number":2,"created":"2019-07-29 10:12:00.000000000","uploader":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"ref":"refs/changes/15/672915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/15/672915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/2"}}},"commit":{"parents":[{"commit":"04b10032b783081303af0431574a805d59bdc2dc","subject":"Disable monasca-tempest-java-influxdb job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/04b10032b783081303af0431574a805d59bdc2dc"}]}],"author":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-07-26 08:35:53.000000000","tz":480},"committer":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-07-29 10:11:22.000000000","tz":480},"subject":"Json loads error","message":"Json loads error\n\nPyhton version below 3.6,function json.loads() throw decoder error.\n3.6 version input option can bytes or bytearray,below 3.6 version\ntype of bytes not cimpatible.\n\nTask: 35991\nStory: 2006285\nChange-Id: Ib1dba51a57fdce028a6b20da06201f0597085047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/0f96fd652cf4fcb97360249c947337313f00ec77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/0f96fd652cf4fcb97360249c947337313f00ec77"}]},"branch":"refs/heads/master"},"22d7c9bb514c9199fce45440ea4b452721ce2a96":{"kind":"REWORK","_number":3,"created":"2019-07-30 00:23:06.000000000","uploader":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"ref":"refs/changes/15/672915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/15/672915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/15/672915/3"}}},"commit":{"parents":[{"commit":"04b10032b783081303af0431574a805d59bdc2dc","subject":"Disable monasca-tempest-java-influxdb job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/04b10032b783081303af0431574a805d59bdc2dc"}]}],"author":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-07-26 08:35:53.000000000","tz":480},"committer":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-07-30 00:15:10.000000000","tz":480},"subject":"Json loads error","message":"Json loads error\n\nPyhton version below 3.6,function json.loads() throw decoder error.\n3.6 version input option can bytes or bytearray,below 3.6 version\ntype of bytes not cimpatible.\n\nTask: 35991\nStory: 2006285\nChange-Id: Ib1dba51a57fdce028a6b20da06201f0597085047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/22d7c9bb514c9199fce45440ea4b452721ce2a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/22d7c9bb514c9199fce45440ea4b452721ce2a96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
