)]}'
{"id":"openstack%2Fmonasca-notification~674146","triplet_id":"openstack%2Fmonasca-notification~master~I8e2775a283b4670a8979541602a3f245c18e5d35","project":"openstack/monasca-notification","branch":"master","topic":"fix-python3-json-loads-bug","hashtags":[],"change_id":"I8e2775a283b4670a8979541602a3f245c18e5d35","subject":"Fix python3 json loads error","status":"MERGED","created":"2019-08-02 00:42:38.000000000","updated":"2019-08-06 16:30:35.000000000","submitted":"2019-08-06 16:30:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674146-1565109035700-a1d2047c","meta_rev_id":"c9cc7117719ee61cda57dcd57f37dc48c80f3d91","_number":674146,"virtual_id_number":674146,"owner":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-06 16:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-05 08:15:54.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-06 16:02:05.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 16:02:05.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-08-06 16:30:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2b35ea1f2d7526f9176fcde9b4c7cc60e30aa62","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-08-02 00:42:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"069afe666f6665c98b508e43b8aa804ba09c5d28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 02:01:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/46/674146/1/check/openstack-tox-cover/3f0f017/cover/ : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints https://logs.opendev.org/46/674146/1/check/openstack-tox-lower-constraints/bba3ea4/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://logs.opendev.org/46/674146/1/check/openstack-tox-pep8/22cb50c/ : SUCCESS in 3m 49s\n- openstack-tox-py27 https://logs.opendev.org/46/674146/1/check/openstack-tox-py27/25a8aa5/ : SUCCESS in 3m 22s\n- openstack-tox-py36 https://logs.opendev.org/46/674146/1/check/openstack-tox-py36/c265460/ : SUCCESS in 3m 09s\n- openstack-tox-py37 https://logs.opendev.org/46/674146/1/check/openstack-tox-py37/e16c48d/ : SUCCESS in 3m 42s\n- build-monasca-docker-image https://logs.opendev.org/46/674146/1/check/build-monasca-docker-image/c7678b8/ : SUCCESS in 6m 40s\n- monasca-tempest-python2-influxdb https://logs.opendev.org/46/674146/1/check/monasca-tempest-python2-influxdb/fd7b88a/ : SUCCESS in 44m 48s (non-voting)\n- monasca-tempest-python3-influxdb https://logs.opendev.org/46/674146/1/check/monasca-tempest-python3-influxdb/a9c8345/ : SUCCESS in 44m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd244962a4251113193d7dcb11fa889d0a9845b8","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-08-02 10:18:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c2b662f9d32b40777f3d6b0086946c96417ff1c","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-08-03 00:58:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7022c608b8af3d7153988dd764f462007f6836bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 01:53:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/46/674146/2/check/openstack-tox-cover/ce65e19/cover/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://logs.opendev.org/46/674146/2/check/openstack-tox-lower-constraints/5785c6c/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://logs.opendev.org/46/674146/2/check/openstack-tox-pep8/499fd3a/ : SUCCESS in 3m 48s\n- openstack-tox-py27 https://logs.opendev.org/46/674146/2/check/openstack-tox-py27/6cf8fbd/ : SUCCESS in 3m 48s\n- openstack-tox-py36 https://logs.opendev.org/46/674146/2/check/openstack-tox-py36/b2bff69/ : SUCCESS in 4m 14s\n- openstack-tox-py37 https://logs.opendev.org/46/674146/2/check/openstack-tox-py37/98089b9/ : SUCCESS in 4m 49s\n- build-monasca-docker-image https://logs.opendev.org/46/674146/2/check/build-monasca-docker-image/5ccd353/ : SUCCESS in 6m 41s\n- monasca-tempest-python2-influxdb https://logs.opendev.org/46/674146/2/check/monasca-tempest-python2-influxdb/37c5840/ : SUCCESS in 51m 18s (non-voting)\n- monasca-tempest-python3-influxdb https://logs.opendev.org/46/674146/2/check/monasca-tempest-python3-influxdb/ae819f3/ : SUCCESS in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d22cfe6f6332ad2193fe6589c6b670d57f521306","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-08-05 08:15:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"27d205af8cb3231e8fb14815655b87d79ee875cd","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-08-06 16:02:05.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e12a9a103257c0e6bda24a3058834c3c264a3dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:02:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"160d811ca4e7712dba4ab53c43a229fdc9712d0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:30:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/46/674146/2/gate/openstack-tox-lower-constraints/20b0322/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://logs.opendev.org/46/674146/2/gate/openstack-tox-pep8/796d7ac/ : SUCCESS in 3m 45s\n- openstack-tox-py27 https://logs.opendev.org/46/674146/2/gate/openstack-tox-py27/8627d8e/ : SUCCESS in 4m 13s\n- openstack-tox-py36 https://logs.opendev.org/46/674146/2/gate/openstack-tox-py36/d370f69/ : SUCCESS in 4m 31s\n- openstack-tox-py37 https://logs.opendev.org/46/674146/2/gate/openstack-tox-py37/fbbbf9a/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a8b398aa16a57fabd866d73f2a41b25d44729674","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:30:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3db7b179ef531e18e5e3b17c54838147ddf3411c","revisions":{"8e3ee830b87d5fb61b1ead1b85411fce128ba2e2":{"kind":"REWORK","_number":1,"created":"2019-08-02 00:42:38.000000000","uploader":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"ref":"refs/changes/46/674146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/46/674146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/46/674146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/46/674146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/46/674146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/46/674146/1"}}},"commit":{"parents":[{"commit":"22d7c9bb514c9199fce45440ea4b452721ce2a96","subject":"Json loads error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/22d7c9bb514c9199fce45440ea4b452721ce2a96"}]}],"author":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-02 00:36:40.000000000","tz":480},"committer":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-02 00:36:40.000000000","tz":480},"subject":"Fix python3 json loads error","message":"Fix python3 json loads error\n\n`oslo_serialization.jsonutils.loads` is a more appropriate solution\nTask: 35991\nStory: 2006285\n\nChange-Id: I8e2775a283b4670a8979541602a3f245c18e5d35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/8e3ee830b87d5fb61b1ead1b85411fce128ba2e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/8e3ee830b87d5fb61b1ead1b85411fce128ba2e2"}]},"branch":"refs/heads/master"},"3db7b179ef531e18e5e3b17c54838147ddf3411c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-03 00:58:03.000000000","uploader":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"ref":"refs/changes/46/674146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/46/674146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/46/674146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/46/674146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/46/674146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/46/674146/2"}}},"commit":{"parents":[{"commit":"22d7c9bb514c9199fce45440ea4b452721ce2a96","subject":"Json loads error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/22d7c9bb514c9199fce45440ea4b452721ce2a96"}]}],"author":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-02 00:36:40.000000000","tz":480},"committer":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-03 00:57:14.000000000","tz":480},"subject":"Fix python3 json loads error","message":"Fix python3 json loads error\n\n`oslo_serialization.jsonutils.loads` is a more appropriate solution\nTask: 36071\nStory: 2006285\n\nChange-Id: I8e2775a283b4670a8979541602a3f245c18e5d35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/3db7b179ef531e18e5e3b17c54838147ddf3411c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/3db7b179ef531e18e5e3b17c54838147ddf3411c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
