)]}'
{"id":"openstack%2Fmonasca-notification~691656","triplet_id":"openstack%2Fmonasca-notification~master~If0649a1fe6a5337a20f0aba9a1a4ea8a0dd04617","project":"openstack/monasca-notification","branch":"master","hashtags":[],"change_id":"If0649a1fe6a5337a20f0aba9a1a4ea8a0dd04617","subject":"Minimal fix for spawning periodic notification process","status":"ABANDONED","created":"2019-10-28 10:57:01.000000000","updated":"2019-11-12 16:32:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e54de04c5b38041d713343c551347ab83785bcb0","_number":691656,"virtual_id_number":691656,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-28 11:43:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":-1,"date":"2019-10-28 11:03:56.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 11:43:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2244990af18c3d990b05250a2b0c4326a50a2257","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-28 10:57:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"744b73dd66043361cc9b8b3b8323cf88a23ad1c7","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-28 10:58:05.000000000","message":"Patch Set 1: Workflow-1\n\nNeed to test","accounts_in_message":[],"_revision_number":1},{"id":"f52959a0631c02fb71634ef46101bf9fd06687bd","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-28 11:03:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9d7bda1391ca12be48e9ba0991617685ab2e0da8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-28 11:03:56.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"484e688c1a548da3d7cef1d8278ce32a29bb2684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 11:43:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8ace63b90514fd2be4eeb3c554f4910 : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/186bd16d922945f086c2ac7f5a5e43ed : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f45bcdfb42f449cb8c37c68c0079b06 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d32b0885fab0416481f3fbfff7d0d33e : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d00ca47fdf1f4d98b833093e15f2079d : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b54c5b7f669495f99c1cedcd9e48596 : SUCCESS in 3m 44s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/d26c41010c4e47499592e0220b3d2e1b : SUCCESS in 6m 54s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/39f3b9bae0c74f0b9a2a529951da4830 : SUCCESS in 36m 21s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/d60379fc05f244579712a131ef9533df : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e54de04c5b38041d713343c551347ab83785bcb0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-12 16:32:01.000000000","message":"Abandoned\n\nSuperseded, see: https://review.opendev.org/#/q/topic:periodic_notifications+(status:open+OR+status:merged)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1daf9ac6aed7a7016f2f3b6979ad9f544400c05e","revisions":{"a896d88dfb240db20786de86f102927424bad3fc":{"kind":"REWORK","_number":1,"created":"2019-10-28 10:57:01.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/56/691656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/56/691656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/56/691656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/56/691656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/56/691656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/56/691656/1"}}},"commit":{"parents":[{"commit":"f062546cba50c5b00b977710625aeb5c22fdb141","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f062546cba50c5b00b977710625aeb5c22fdb141"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-10-28 10:52:52.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-10-28 10:56:03.000000000","tz":0},"subject":"Minimal fix for spawning periodic notification process","message":"Minimal fix for spawning periodic notification process\n\nThe current comparision between an int and a dict always returns\nfalse so the periodic notification process never spawns. This minimal\nfix should get it working again. A later patch could support user\ndefined periods.\n\nChange-Id: If0649a1fe6a5337a20f0aba9a1a4ea8a0dd04617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/a896d88dfb240db20786de86f102927424bad3fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/a896d88dfb240db20786de86f102927424bad3fc"}]},"branch":"refs/heads/master"},"1daf9ac6aed7a7016f2f3b6979ad9f544400c05e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-28 11:03:45.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/56/691656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/56/691656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/56/691656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/56/691656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/56/691656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/56/691656/2"}}},"commit":{"parents":[{"commit":"f062546cba50c5b00b977710625aeb5c22fdb141","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f062546cba50c5b00b977710625aeb5c22fdb141"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-10-28 10:52:52.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-10-28 11:03:23.000000000","tz":0},"subject":"Minimal fix for spawning periodic notification process","message":"Minimal fix for spawning periodic notification process\n\nThe current comparision between an int and a dict always returns\nfalse so the periodic notification process never spawns. This minimal\nfix should get it working again. A later patch could support user\ndefined periods.\n\nStory: 2006783\nTask: 37313\nChange-Id: If0649a1fe6a5337a20f0aba9a1a4ea8a0dd04617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/1daf9ac6aed7a7016f2f3b6979ad9f544400c05e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/1daf9ac6aed7a7016f2f3b6979ad9f544400c05e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
