)]}'
{"id":"openstack%2Fmonasca-notification~693860","triplet_id":"openstack%2Fmonasca-notification~master~Icd1b12f4b1e452f8285b8402f49815dd7f72b1ea","project":"openstack/monasca-notification","branch":"master","topic":"periodic_notifications","hashtags":[],"change_id":"Icd1b12f4b1e452f8285b8402f49815dd7f72b1ea","subject":"Allow users to set periodic notifications on all notification types","status":"MERGED","created":"2019-11-12 15:12:57.000000000","updated":"2019-12-12 16:05:45.000000000","submitted":"2019-12-12 16:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693860-1576166608115-8b70566f","meta_rev_id":"f4bc825c5d538306be8cf2112076c2508714185c","_number":693860,"virtual_id_number":693860,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 16:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-11 12:53:38.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-12 15:48:35.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 15:48:35.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-12-12 16:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a018a5f0222e7875e211b511ecf0816b886c2a43","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-12 15:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ad27ec4362fbbf8e2aab6f4fd77b69fbcf36970","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-12 15:18:31.000000000","message":"Topic set to periodic_notifications","accounts_in_message":[],"_revision_number":1},{"id":"0cba3ca8643705f228643e557535a5a152f727c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 15:52:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbdac558ce37409f9e638ee062993be7 : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d232207c6fa425e896ef9d0fc004a5b : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108696c5f6f741efaee4e5f8ca866240 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7f3f8aa51af4db280180eba67784cfe : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47c380470bc74fca9dc45cbdff6f3944 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91d274270c444e0b8ac1abaee87d5579 : SUCCESS in 4m 20s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/1fc132d22d4d4ccf93f61a2ede965dd5 : SUCCESS in 8m 13s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/4e9361564768469ab9378e3fea255c2b : SUCCESS in 38m 00s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ebcc7833acf84da999c7b43448638e5c : SUCCESS in 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ace0681fcdc11c8960363f1aed325f4c4396c4f0","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-12-11 10:53:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2aba2849d650f933b48dcfbf7fa7b0d72b9beb1d","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-12-11 11:04:00.000000000","message":"Patch Set 1: -Code-Review\n\nPlease add the release note.","accounts_in_message":[],"_revision_number":1},{"id":"bbc897480b993ac64dad368fc9fe1bdb79bb495a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-11 11:14:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8343451dbdb17f545b82eb759eb92f521b2d375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 11:54:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b2661426ad54893b0d7c807b311d7bf : SUCCESS in 3m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/669767bdfe7a4762a9d384d6a073ed29 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001bb203303e44d3918b84a49e3add73 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57bbecf4270b478ca94fab6470c0e849 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5995e38aca24383b71d863172455e6a : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e494bd3150eb496fb1b9e1efe2c563f9 : SUCCESS in 3m 32s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/45d5c079b0cb420ab46ea80a6f733d45 : SUCCESS in 8m 33s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e5775f5aaaf04eb2b3cb1a31c0ce0f8e : SUCCESS in 38m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"840b289d3979a677248b884d851f521da121e328","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-12-11 12:53:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c898dc40b08daf02d1cd4b42e6a6863f357a43b9","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-12-12 15:48:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3a3d7bf0d67b35dcfbebb47891777f589d08fefc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:48:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"39caf9a4aaeb57d7c3d7a59c0f4863b5c6ab435f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 16:03:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a750332141c4a739d25c3a180386af6 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408c225a09784a5385b481d4849dedf2 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b72d10cd04cf44f9808dc07204fadb9a : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f2cd9dd537b42ac8c7bd7569cb02a4d : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68fb07ae314f4f7dbf82b753846f45ae : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d50347430488a18c0d06d52790dce1169d3d5587","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 16:03:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f4bc825c5d538306be8cf2112076c2508714185c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 16:05:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61c8cefda96b4ec79dfa54abe53c7e3d : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"044b9c7d1175d775d92503c613175ebcae27a718","revisions":{"732252f440570da3f6cb4b6ca7d03e3f83116560":{"kind":"REWORK","_number":1,"created":"2019-11-12 15:12:57.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/60/693860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/60/693860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/693860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/693860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/693860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/60/693860/1"}}},"commit":{"parents":[{"commit":"8f0902e7ab718dc9232c5497eaad6be45144a247","subject":"Fix periodic notifications for webhooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/8f0902e7ab718dc9232c5497eaad6be45144a247"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-11-12 10:52:40.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-11-12 15:12:24.000000000","tz":0},"subject":"Allow users to set periodic notifications on all notification types","message":"Allow users to set periodic notifications on all notification types\n\nIn some cases, users may want to send periodic notifications for\nnotification types other than webhooks.\n\nStory: 2006837\nTask: 37412\nChange-Id: Icd1b12f4b1e452f8285b8402f49815dd7f72b1ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/732252f440570da3f6cb4b6ca7d03e3f83116560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/732252f440570da3f6cb4b6ca7d03e3f83116560"}]},"branch":"refs/heads/master"},"044b9c7d1175d775d92503c613175ebcae27a718":{"kind":"REWORK","_number":2,"created":"2019-12-11 11:14:14.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/60/693860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/60/693860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/693860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/693860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/60/693860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/60/693860/2"}}},"commit":{"parents":[{"commit":"efc6e28edceac26599eab7f274a40708d1889f8c","subject":"Fix periodic notifications for webhooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/efc6e28edceac26599eab7f274a40708d1889f8c"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-11-12 10:52:40.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-11 11:13:25.000000000","tz":0},"subject":"Allow users to set periodic notifications on all notification types","message":"Allow users to set periodic notifications on all notification types\n\nIn some cases, users may want to send periodic notifications for\nnotification types other than webhooks.\n\nStory: 2006837\nTask: 37412\nChange-Id: Icd1b12f4b1e452f8285b8402f49815dd7f72b1ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/044b9c7d1175d775d92503c613175ebcae27a718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/044b9c7d1175d775d92503c613175ebcae27a718"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
