)]}'
{"id":"openstack%2Fmonasca-notification~709450","triplet_id":"openstack%2Fmonasca-notification~master~I327931591e72efd8890e5bb6bbe1cc6a712cb193","project":"openstack/monasca-notification","branch":"master","hashtags":[],"change_id":"I327931591e72efd8890e5bb6bbe1cc6a712cb193","subject":"Fix program name CLI usage info","status":"ABANDONED","created":"2020-02-24 11:23:55.000000000","updated":"2026-02-03 22:36:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b40b20ec0ca2ea8e83276e302d0b54144672fac1","_number":709450,"virtual_id_number":709450,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-24 14:51:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2020-02-24 17:33:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 11:25:13.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-02-24 14:51:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-24 17:33:57.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2026-02-03 22:36:06.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"ace66ddea930b4a53731806cd2ebd59ce4e3197b","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-24 11:23:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6455cca4d8071e0c492c309de404e8f0aaa78350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 12:02:36.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"efee89dd625d7965c2071570c73e01a5b623e3cd","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-24 12:49:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e91a81a6d0af4c3bfd9c48d97ab83ab6579df69c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 13:42:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eae15f18cc4c4e4783e331b9e27e3beb : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71df1e14e266445696a1a69cbabedd20 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4501f86522c54785b5477effc47aa872 : FAILURE in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b767ff886fb4f9689ce9c526942610d : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ff397dbc2584abd83cb9e13852ce90c : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb19939317b94371bf5622b43f10d087 : SUCCESS in 6m 10s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/1e6a7374c15e4b9a97cd2bb0d55b25e1 : SUCCESS in 9m 14s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ab3f1d3958294bcc99e1d36d4d71128d : FAILURE in 48m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c00958d36d99f598595a7ae0b8259c9f8b1e415b","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-24 13:50:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe4b0230f31c09823f3f85bb0decb2e605139b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:51:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b4761170a8f41b5b297e502bf4dbbe5 : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9af7ed60af5a416db2a723394990ff43 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607498eae2bd416aace92a160e1b5c4c : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0da123e7e384a4aac6e020203184e28 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b205dfa27174db894b7168414bef9a4 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d7b3e01d64347dc8b6a3770dce8779b : SUCCESS in 5m 58s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/1dcdf329e2234077beaf93d4651cc48f : SUCCESS in 10m 23s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/55799b4ffd824f5cbc3c129e9e7efbb2 : SUCCESS in 48m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9b5cc309d9f14259e7a9c1851eb2f1e16ba8a79d","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-02-24 17:33:57.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks more reliable.","accounts_in_message":[],"_revision_number":3},{"id":"b40b20ec0ca2ea8e83276e302d0b54144672fac1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:36:06.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d8f459fca54538f85b321592669e201ea0f8b52f","revisions":{"0074211f0656507895799f3130cbb8e9f984a40b":{"kind":"REWORK","_number":1,"created":"2020-02-24 11:23:55.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/50/709450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/50/709450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/1"}}},"commit":{"parents":[{"commit":"76f86ec6123f0ef3aaa75200253939340c5d03d5","subject":"Declare cli.yaml_config option as not required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/76f86ec6123f0ef3aaa75200253939340c5d03d5"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-24 11:21:35.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-24 11:21:35.000000000","tz":60},"subject":"Fix program name CLI usage info","message":"Fix program name CLI usage info\n\nStory: 2007332\nTask: 38856\n\nChange-Id: I327931591e72efd8890e5bb6bbe1cc6a712cb193\nCo-authored-by: ricolin \u003crico.lin.guanyu@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/0074211f0656507895799f3130cbb8e9f984a40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/0074211f0656507895799f3130cbb8e9f984a40b"}]},"branch":"refs/heads/master"},"6c8ec9ea76623c60b112f1c0a21fd8f5aef600a6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-24 12:49:47.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/50/709450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/50/709450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/2"}}},"commit":{"parents":[{"commit":"76f86ec6123f0ef3aaa75200253939340c5d03d5","subject":"Declare cli.yaml_config option as not required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/76f86ec6123f0ef3aaa75200253939340c5d03d5"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-24 11:21:35.000000000","tz":60},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-24 12:49:44.000000000","tz":0},"subject":"Fix program name in CLI usage info","message":"Fix program name in CLI usage info\n\nStory: 2007332\nTask: 38856\n\nChange-Id: I327931591e72efd8890e5bb6bbe1cc6a712cb193\nCo-authored-by: ricolin \u003crico.lin.guanyu@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/6c8ec9ea76623c60b112f1c0a21fd8f5aef600a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/6c8ec9ea76623c60b112f1c0a21fd8f5aef600a6"}]},"branch":"refs/heads/master"},"d8f459fca54538f85b321592669e201ea0f8b52f":{"kind":"REWORK","_number":3,"created":"2020-02-24 13:50:07.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/50/709450/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/50/709450/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/50/709450/3"}}},"commit":{"parents":[{"commit":"76f86ec6123f0ef3aaa75200253939340c5d03d5","subject":"Declare cli.yaml_config option as not required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/76f86ec6123f0ef3aaa75200253939340c5d03d5"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-24 11:21:35.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-24 13:48:34.000000000","tz":60},"subject":"Fix program name CLI usage info","message":"Fix program name CLI usage info\n\nStory: 2007332\nTask: 38856\n\nChange-Id: I327931591e72efd8890e5bb6bbe1cc6a712cb193\nCo-authored-by: ricolin \u003crico.lin.guanyu@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/d8f459fca54538f85b321592669e201ea0f8b52f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/d8f459fca54538f85b321592669e201ea0f8b52f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
