)]}'
{"id":"openstack%2Fmonasca-notification~730857","triplet_id":"openstack%2Fmonasca-notification~master~Ia1c223e51c9f82b8b83ebd187b6672296f8e6105","project":"openstack/monasca-notification","branch":"master","topic":"ujson-replacement","hashtags":[],"change_id":"Ia1c223e51c9f82b8b83ebd187b6672296f8e6105","subject":"Remove ujson from lower-constraints","status":"MERGED","created":"2020-05-26 15:39:51.000000000","updated":"2020-05-26 19:02:21.000000000","submitted":"2020-05-26 19:02:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730857-1590519741377-41f9ad85","meta_rev_id":"3a24b3e7fb8f22f8bf0caee804cff9c707104012","_number":730857,"virtual_id_number":730857,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 19:02:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-26 18:53:16.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-26 18:53:16.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 18:53:16.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-05-26 19:02:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"893fab2b44cee676a37989286374fd8c2a0ba202","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-26 15:39:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2173799f489218832267d64eeb6380f97f0d0bcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:28:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e240422f0e34dd7b7ab717fc57dcf46 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/965c1131b3164140b1894b9bdf0ee8bd : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e329d158cc94ead86ff318af696e45b : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1210e0cae9cd47ffa02939f9359215e0 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d3d3a1b0ed14c08b0c48244359f1b07 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26859b99daac4ffeafbc401bc8c703d9 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cfd3eff39e741b790866a33d1f08a56 : SUCCESS in 5m 19s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/0d6daba2551f4913861c6b58443df669 : SUCCESS in 9m 27s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/c61248dc9be940c78252fbc0a12cb21a : SUCCESS in 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c0172325822914ae3b1ea0a5ce00fd11ad48a8d","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-26 18:53:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ff9d5b1d951fd71a7fbe8eb4b79e8c9bcfe30e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:53:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a7269172b9f883638a047ee0b817e49064bbf36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 19:02:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"51fce3b07267b9e9292a3a0935c01d88acf8603d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 19:02:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/39f891008b2542ab83df2c0967d0bb1b : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f85df43a5ca74c1684a748c47b3ec6e2 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b867b7c2be1842f58f10e0bef22489c0 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4534b0c8f5464e20864a71c2b41a3751 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f922124a172a4cf3a6ab248c454fe171 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5b0cbe021317294e458b4363dc8514b38dbb1e9","revisions":{"a5b0cbe021317294e458b4363dc8514b38dbb1e9":{"kind":"REWORK","_number":1,"created":"2020-05-26 15:39:51.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/57/730857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/57/730857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/57/730857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/57/730857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/57/730857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/57/730857/1"}}},"commit":{"parents":[{"commit":"f0e2d44759588cc1957ce37d1b87859f7bfdbf7a","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f0e2d44759588cc1957ce37d1b87859f7bfdbf7a"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-05-26 15:39:09.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-05-26 15:39:09.000000000","tz":120},"subject":"Remove ujson from lower-constraints","message":"Remove ujson from lower-constraints\n\nChange-Id: Ia1c223e51c9f82b8b83ebd187b6672296f8e6105\nStory: 2007549\nTask: 39859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/a5b0cbe021317294e458b4363dc8514b38dbb1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/a5b0cbe021317294e458b4363dc8514b38dbb1e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
