)]}'
{"id":"openstack%2Fmonasca-notification~734379","triplet_id":"openstack%2Fmonasca-notification~master~Ib7c9cca06dcfe14440ab9711cc0b76a5b4fe4e1c","project":"openstack/monasca-notification","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"Ib7c9cca06dcfe14440ab9711cc0b76a5b4fe4e1c","subject":"Use unittest.mock instead of mock","status":"MERGED","created":"2020-06-08 23:25:15.000000000","updated":"2020-06-10 11:04:56.000000000","submitted":"2020-06-10 11:04:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734379-1591787096862-f90b450b","meta_rev_id":"b8dbd333f0e5de11116e6b7121b2fb6b88887032","_number":734379,"virtual_id_number":734379,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 11:04:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-10 10:57:12.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-10 10:57:12.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 10:57:12.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-06-10 11:04:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55e7c0d4635d59fe2e26371f672ca0bbba91d884","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 23:25:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bee11a28610fb7b67e123c7e7372f678b191404e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 00:02:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b72211f842224aa7a0ef8e7f93753748 : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/053856c1282e46a88f3d7baca3896613 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72e6c9295b224ca0946a8fad64355f92 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b7c10ff6f84b0ba0d6296bbb346ef8 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7efefcb62168404aaacdf134e9c12b6e : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de68f24615ee48fc9eb22eb9fe5c7237 : SUCCESS in 7m 36s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/bdff80b3730a4e57af0851fc6c35c037 : SUCCESS in 9m 36s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/463d3d10b6094b10b1561baa48f774cc : SUCCESS in 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51981b0b374807f8c8d29ee2d7a435d467a4e54c","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-06-10 10:57:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5c329ba28aa8e129e13a01eb5713d155239a502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 10:57:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2eb8a26ea21e78c93c0d37c5af46e9c97fe5d7bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 11:04:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"78a6852ade52898200e7a2a49bb73c9ee5164540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 11:04:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/944663394331446a8cd51cd7654bcdf1 : SUCCESS in 2m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9488e00d9cc340198ab2ad32dbd9d14c : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4ce3562b5a4ba78cf3ba8a477599ca : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a515043385744ae59e39da905b850a6b : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6b1eedcd754448fa711efef9d5ea3e9 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7faa2184cd0e2cd9c95785431984bf9cb1b6e987","revisions":{"7faa2184cd0e2cd9c95785431984bf9cb1b6e987":{"kind":"REWORK","_number":1,"created":"2020-06-08 23:25:15.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/79/734379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/79/734379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/79/734379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/79/734379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/79/734379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/79/734379/1"}}},"commit":{"parents":[{"commit":"a49af326e5022e15a1b5ea0752536a5e6bedd9c7","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/a49af326e5022e15a1b5ea0752536a5e6bedd9c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:25:09.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:25:09.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: Ib7c9cca06dcfe14440ab9711cc0b76a5b4fe4e1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/7faa2184cd0e2cd9c95785431984bf9cb1b6e987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/7faa2184cd0e2cd9c95785431984bf9cb1b6e987"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
