)]}'
{"id":"openstack%2Fmonasca-notification~754966","triplet_id":"openstack%2Fmonasca-notification~master~I736f141b53d5fb985e5deaa5d77c552ebad6fccd","project":"openstack/monasca-notification","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I736f141b53d5fb985e5deaa5d77c552ebad6fccd","subject":"[goal] Migrate testing to ubuntu focal As per victoria cycle testing runtime and community goal[1] we need to migrate upstream CI/CD to Ubuntu Focal(20.04).","status":"MERGED","created":"2020-09-29 10:29:51.000000000","updated":"2020-09-30 09:51:14.000000000","submitted":"2020-09-30 09:08:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754966-1601456932490-fe0e020a","meta_rev_id":"5d2d1c2317aaaa7c33af065e3c9c0c1d92ba0a8e","_number":754966,"virtual_id_number":754966,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 09:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"date":"2020-09-30 07:39:37.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-30 08:56:58.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-30 08:56:58.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 20:10:53.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-30 08:56:58.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-09-30 09:08:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e8cb740730925cc8eab09556b58f321365ee70c","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-09-29 10:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61b095b7225edfc35d780840004912987e5c628b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 11:05:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d73412da69124485a6c51517755b3b67 : FAILURE in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e710139d76046958a83889902eee43a : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/495f84756c844a29a72cf420a6a24417 : FAILURE in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b90dee27e88648ac85b4cec040723298 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/879608f039d849e980f5bbb6dce1b788 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8902d23d9d4046cfabcfc93da7b2422a : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ae4f13f663b45de8d1429507717313e : SUCCESS in 3m 24s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/c2b44f97595b4e2a8214e89f3c6e7d55 : SUCCESS in 7m 55s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/3d66d2ba756442ba9106f2d184eff87f : FAILURE in 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d425d4909a5ccc7e1bc89ce7dcc453489abb01cb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-29 20:10:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d76bf82e32770639e9ea64d0091be75677e5d723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 21:12:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/21826e5341f44ee199a9d0f3253f52c2 : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/129ada5af7784baab26849bf160fd786 : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61ea79d7dacb4c469babe3ae79ca4555 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa24a75f908d4424b4f759660a2e793b : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1216972f3a92441c916d26dfd487423f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc46531ff01d479f857fc0dbd8a97a71 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1560ff452ea40d6abd8349d167f5199 : SUCCESS in 3m 39s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/d1ba0ffb596e40378d28d8a83e128673 : SUCCESS in 8m 25s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/44827c02b2a146f890685a5d3c7f2cba : FAILURE in 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7c06cfd85d98e3bcd32de4650c865d239582dd59","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-09-30 07:39:37.000000000","message":"Patch Set 2:\n\nThanks for the help Ghanshyam","accounts_in_message":[],"_revision_number":2},{"id":"d7c3323e9e4acbbd72f14228821f22de15a6595d","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-09-30 08:56:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"352e0bab5d15b1896aea5ead48d8091fc2e72648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 08:57:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c116114d1b939ddd340ee6cb707f6962cd9dc569","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 09:08:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c27276f19f01432c9383f7839de74fea : SUCCESS in 2m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ebb7ce3873d432f80a012f6d0e4fdab : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b760d211a15842cf8abc0e32b54f3b0c : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44b6a78831014cfaa704321a8aba6afc : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9baa7dacf51c46258293b21ffa56d4cb : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81aa3fc1203b4046a767278bc51fb762 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9a00e83dfd0d7555116166e37275c12469131910","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 09:08:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"62ca9bb10002c80a13f1a032a41adb5d0c01c9fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 09:10:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4de3637c791410b8e3cc96f64e1487a : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2},{"id":"becd2abd1c0a53671f91ebfa37e6c37db1678f99","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-09-30 09:51:14.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit fc7ef13781c72cbf3383d1c6dc5391941a0673ae","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44fe8f1c39d9f9ce8eb66c204ce813873c3cc4ae","revisions":{"6a6895271ad9351b86c7fe5b8c16fad94c6471f4":{"kind":"REWORK","_number":1,"created":"2020-09-29 10:29:51.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/66/754966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/66/754966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/66/754966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/66/754966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/66/754966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/66/754966/1"}}},"commit":{"parents":[{"commit":"f9978670b8437b106cc8a4fa88f6760da98c3b87","subject":"Set legacy_kafka_client_enabled \u003d False on default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f9978670b8437b106cc8a4fa88f6760da98c3b87"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-09-29 10:25:51.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-09-29 10:25:51.000000000","tz":120},"subject":"[goal] Migrate testing to ubuntu focal As per victoria cycle testing runtime and community goal[1] we need to migrate upstream CI/CD to Ubuntu Focal(20.04).","message":"[goal] Migrate testing to ubuntu focal\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nBump netaddr to 0.7.20 in lower-constraints.txt\nsince it is the min version with py3.8 support [2]\n\nBump psycopg2 to 2.8.4 in lower-constraints.txt\nsince it is the min version with py3.8 support [3]\n\nStory: #2007865\nTask: #40197\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n[2] https://github.com/netaddr/netaddr/commit/cc2a9bb31fae00f18fcdbd97a29b643581fc2a45\n[3] https://pypi.org/project/psycopg2/2.8.4/\n\nChange-Id: I736f141b53d5fb985e5deaa5d77c552ebad6fccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/6a6895271ad9351b86c7fe5b8c16fad94c6471f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/6a6895271ad9351b86c7fe5b8c16fad94c6471f4"}]},"branch":"refs/heads/master"},"44fe8f1c39d9f9ce8eb66c204ce813873c3cc4ae":{"kind":"REWORK","_number":2,"created":"2020-09-29 20:10:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/754966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/66/754966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/66/754966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/66/754966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/66/754966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/66/754966/2"}}},"commit":{"parents":[{"commit":"f9978670b8437b106cc8a4fa88f6760da98c3b87","subject":"Set legacy_kafka_client_enabled \u003d False on default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/f9978670b8437b106cc8a4fa88f6760da98c3b87"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-09-29 10:25:51.000000000","tz":120},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-29 20:10:52.000000000","tz":-300},"subject":"[goal] Migrate testing to ubuntu focal As per victoria cycle testing runtime and community goal[1] we need to migrate upstream CI/CD to Ubuntu Focal(20.04).","message":"[goal] Migrate testing to ubuntu focal\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nBump netaddr to 0.7.20 in lower-constraints.txt\nsince it is the min version with py3.8 support [2]\n\nBump psycopg2 to 2.8.4 in lower-constraints.txt\nsince it is the min version with py3.8 support [3]\n\nStory: #2007865\nTask: #40197\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n[2] https://github.com/netaddr/netaddr/commit/cc2a9bb31fae00f18fcdbd97a29b643581fc2a45\n[3] https://pypi.org/project/psycopg2/2.8.4/\n\nChange-Id: I736f141b53d5fb985e5deaa5d77c552ebad6fccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/44fe8f1c39d9f9ce8eb66c204ce813873c3cc4ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/44fe8f1c39d9f9ce8eb66c204ce813873c3cc4ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
