)]}'
{"id":"openstack%2Fmonasca-notification~757901","triplet_id":"openstack%2Fmonasca-notification~master~I6892059b122cd5ec84edd41fc55a6aa2ea6267a9","project":"openstack/monasca-notification","branch":"master","hashtags":[],"change_id":"I6892059b122cd5ec84edd41fc55a6aa2ea6267a9","subject":"Set Docker publish jobs","status":"MERGED","created":"2020-10-13 15:49:54.000000000","updated":"2020-10-15 09:28:06.000000000","submitted":"2020-10-14 13:52:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757901-1602683566970-332a2ac9","meta_rev_id":"e6e7e0667079fd8ea95ecb1f7c980dde7b75667c","_number":757901,"virtual_id_number":757901,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 13:52:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":2,"date":"2020-10-14 13:34:15.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":1,"date":"2020-10-14 13:34:15.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 16:42:20.000000000","updated_by":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"reviewer":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"state":"REVIEWER"},{"updated":"2020-10-13 16:42:35.000000000","updated_by":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"reviewer":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"state":"REVIEWER"},{"updated":"2020-10-14 13:34:15.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-10-14 13:52:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71627ecdfebb9714343ad617c06fdcf8057d0ea0","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-13 15:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"999ce839de63e964a31d22de8f3e65cfe4e1b444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:39:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea03a780d49846ef9374ee15e7ce02dd : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/138d5f65709e4e21af89ed89b6d75482 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5c9c65f60b64571a8beedbf5c6234d6 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/509f088a8404461d9f5b9574fe024f5d : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f24df6358044a3285abd1bd1cefab1c : SUCCESS in 5m 55s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/1ef6eb6e672f444f9d04a2244e9f3c94 : SUCCESS in 7m 54s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/3b72d021bf2a4550aab2a5567003f9e2 : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c6578cf3d6b672e5dea589c84cb6b070bea745c5","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-10-14 10:39:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e0cd0828ae238299129bbcd3f74048a49bd5eca9","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-14 12:10:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0fefa9fbfa112424f725c7846dd62e2d22f75319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:58:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfba28c463a044cd9d7ec0a878baade0 : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c076314effc54d63b7a4a61b5e0bd52f : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2269a847011445a6a2d078cab5137f3b : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d2c62bb881c4b90b938f567a43566bb : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e460126a5f548258ea87278030ae2e5 : SUCCESS in 5m 52s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/665779d0b5c04d76a09d46422cd94f99 : SUCCESS in 10m 28s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/c38e33b23d474863b12486b46f2fa45a : SUCCESS in 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4778284511ff761e91121e23bb73a41287efe58e","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-10-14 13:34:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7892eb6171301ce2aeb37a710537a4b4d66029f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:34:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a3b58689ab5e8d4015c1388235613a78b1177b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:52:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25abf0560f8543aeb551d975ddbf6954 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce243c589d54f2b91e4d0b169c7c567 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a567b5b8a6439390730eb95a7df1fb : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5451bd584ce4fb7ae330581e25c15cd : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e595ad993ed07494aa881e4611dce2e8c2f70661","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:52:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5b73762183028886de725d715261e911ed111817","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-14 14:13:44.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 84772ba3d0808a796f6db2ac5acf078eed815144","accounts_in_message":[],"_revision_number":2},{"id":"e6e7e0667079fd8ea95ecb1f7c980dde7b75667c","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-15 09:28:06.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 88df8d2dc140eff3851246c33f1046ced2cc5cd0","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"31da3058812d8a8eb9ef77f03b0e94f19a5e34b1","revisions":{"d084ffba1d97f5f3636316cf0ebead8a9608512e":{"kind":"REWORK","_number":1,"created":"2020-10-13 15:49:54.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/01/757901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/01/757901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/01/757901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/01/757901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/01/757901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/01/757901/1"}}},"commit":{"parents":[{"commit":"e8c1e33c1672df64e2645a17c55a8ec8dd1686d1","subject":"Remove deprecated YAML config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e8c1e33c1672df64e2645a17c55a8ec8dd1686d1"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-10-13 15:40:49.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-10-13 15:49:28.000000000","tz":120},"subject":"Set Docker publish jobs","message":"Set Docker publish jobs\n\n- Set as voting:\nbuild-monasca-docker-image\nmonasca-tempest-python3-influxdb\n\n- Bringing back the Docker publish jobs which were temporarily\ndisabled [1]\n\n[1] https://opendev.org/openstack/monasca-notification/commit/db01fe425bc6cf986ff6cd3695977c15b803bc06\n\nChange-Id: I6892059b122cd5ec84edd41fc55a6aa2ea6267a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/d084ffba1d97f5f3636316cf0ebead8a9608512e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/d084ffba1d97f5f3636316cf0ebead8a9608512e"}]},"branch":"refs/heads/master"},"31da3058812d8a8eb9ef77f03b0e94f19a5e34b1":{"kind":"REWORK","_number":2,"created":"2020-10-14 12:10:55.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/01/757901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-notification","ref":"refs/changes/01/757901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/01/757901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/01/757901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-notification refs/changes/01/757901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-notification refs/changes/01/757901/2"}}},"commit":{"parents":[{"commit":"e8c1e33c1672df64e2645a17c55a8ec8dd1686d1","subject":"Remove deprecated YAML config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/e8c1e33c1672df64e2645a17c55a8ec8dd1686d1"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-10-13 15:40:49.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-10-14 12:10:04.000000000","tz":120},"subject":"Set Docker publish jobs","message":"Set Docker publish jobs\n\n- Set as voting:\nbuild-monasca-docker-image\n\n- Bringing back the Docker publish jobs which were temporarily\ndisabled [1]\n\n[1] https://opendev.org/openstack/monasca-notification/commit/db01fe425bc6cf986ff6cd3695977c15b803bc06\n\nChange-Id: I6892059b122cd5ec84edd41fc55a6aa2ea6267a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/31da3058812d8a8eb9ef77f03b0e94f19a5e34b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-notification/commit/31da3058812d8a8eb9ef77f03b0e94f19a5e34b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
